)]}'
{"id":"openstack%2Fkeystone-specs~698950","triplet_id":"openstack%2Fkeystone-specs~master~I6872b67a254c12056c4484b53a5647618c37916d","project":"openstack/keystone-specs","branch":"master","topic":"bug/1816076","hashtags":[],"change_id":"I6872b67a254c12056c4484b53a5647618c37916d","subject":"Repropose federated attributes in the user API for Ussuri","status":"MERGED","created":"2019-12-13 16:20:27.000000000","updated":"2020-01-28 17:16:24.000000000","submitted":"2020-01-28 17:14:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698950-1580231688103-bbef5f2b","meta_rev_id":"de7de18c5220536b296b8326cf0dddec266678a7","_number":698950,"virtual_id_number":698950,"owner":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-28 17:14:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-12-23 18:52:35.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2020-01-24 05:08:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-28 17:07:21.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-28 17:07:21.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-23 18:52:35.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-01-24 05:08:21.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2020-01-28 17:07:21.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-01-28 17:14:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"919fd3ec801819682308db1c2f9b74f8f2468918","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2019-12-13 16:20:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfbe94109ac8ec36980b36017d4ce316ff93e13e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 16:28:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc4257165314484a9bd3b642834d48b1 : FAILURE in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8b0c8d0d97381916cf3b716debc6ed302e5f359a","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2019-12-17 15:12:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a5e14c62ef62905bc1e67c71b220dfd1a3939ae9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 15:21:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec9d85f9749b49e2b035714aaffbf31b : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"c0847473b03f186918ceca97a287e8dd5c42d7bf","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-12-23 18:52:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ba66226a275a48ebe2d0125f3aba7aa1e45c06f5","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2020-01-24 05:08:21.000000000","message":"Patch Set 2: Code-Review+1\n\nThis is great if the user can create a federated user. Looking forward for it.","accounts_in_message":[],"_revision_number":2},{"id":"5e1cf30ebda6bea1c7c905461ddd2620a97d08fb","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-01-28 17:07:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fcbff7ceaf0fffb29b115bda2092ee18a8275bc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 17:07:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ae6fda7d88850f42c36d91d16c4aa32e11cd75ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 17:14:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27ee7d28b0df47bca19a2346d3d4beae : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b056c1f7adba0d5b9169cda51adc0ae8765914d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 17:14:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"55a9b2f4bc8b89ee3259c06405c202fe633f8f36","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 17:16:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/974a162f0c9e489683826cd0c3b2cba4 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"569101e9ab51ccb406a93f0d101c7b4c402e4f18","revisions":{"4f591b04b8a7daf459986c103bb92101e7c4bd81":{"kind":"REWORK","_number":1,"created":"2019-12-13 16:20:27.000000000","uploader":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"ref":"refs/changes/50/698950/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/50/698950/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/50/698950/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/50/698950/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/50/698950/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/50/698950/1"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2019-12-13 16:19:16.000000000","tz":-300},"committer":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2019-12-13 16:19:16.000000000","tz":-300},"subject":"Repropose federated attributes in the user API for Ussuri","message":"Repropose federated attributes in the user API for Ussuri\n\nChange-Id: I6872b67a254c12056c4484b53a5647618c37916d\nRelated-Bug: 1816076\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/4f591b04b8a7daf459986c103bb92101e7c4bd81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/4f591b04b8a7daf459986c103bb92101e7c4bd81"}]},"branch":"refs/heads/master"},"569101e9ab51ccb406a93f0d101c7b4c402e4f18":{"kind":"REWORK","_number":2,"created":"2019-12-17 15:12:15.000000000","uploader":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"ref":"refs/changes/50/698950/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/50/698950/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/50/698950/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/50/698950/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/50/698950/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/50/698950/2"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2019-12-13 16:19:16.000000000","tz":-300},"committer":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2019-12-17 15:12:08.000000000","tz":-300},"subject":"Repropose federated attributes in the user API for Ussuri","message":"Repropose federated attributes in the user API for Ussuri\n\nChange-Id: I6872b67a254c12056c4484b53a5647618c37916d\nRelated-Bug: 1816076\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/569101e9ab51ccb406a93f0d101c7b4c402e4f18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/569101e9ab51ccb406a93f0d101c7b4c402e4f18"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
