)]}'
{"id":"openstack%2Fkeystone-specs~698951","triplet_id":"openstack%2Fkeystone-specs~master~Ie133c14ffba5e4189265920759bfb5e1391f1189","project":"openstack/keystone-specs","branch":"master","topic":"bug/1809116","hashtags":[],"change_id":"Ie133c14ffba5e4189265920759bfb5e1391f1189","subject":"Repropose Expiring Group Membership for Ussuri","status":"MERGED","created":"2019-12-13 16:22:58.000000000","updated":"2020-01-28 17:18:26.000000000","submitted":"2020-01-28 17:16:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698951-1580231813673-cc57de1a","meta_rev_id":"0a6aa19d9ccb4fb2691d392ae395d21be3c29315","_number":698951,"virtual_id_number":698951,"owner":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-28 17:16:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-21 22:44:52.000000000","_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-12-23 18:53:08.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":2,"date":"2020-01-28 17:07:40.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"value":1,"date":"2020-01-28 17:07:40.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-23 18:53:08.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-01-21 22:44:52.000000000","updated_by":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"reviewer":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"state":"REVIEWER"},{"updated":"2020-01-28 17:07:40.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2020-01-28 17:16:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a7dd910b38365850bd1d23f8dc87e0895cc2f950","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2019-12-13 16:22:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a32ec3094ad9f9ce38db2d26b41abf9c1df4755","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 16:36:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8222deb0ce84eb6a3378415c731881b : FAILURE in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"e7ed677bbf1333a0724c6d9f17b7ac93b69d2ca3","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2019-12-17 15:09:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2d507fd9d2172a444fa3cc61e4d31d01ac093012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 15:19:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0ab0753226043cb8303b000a2f2d2e0 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"f79d5adb1e70c3cc2861ce3f2a480875b3744425","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-12-23 18:53:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0acdea55d7e9b33813ba57e38847c1738a01d275","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-01-21 22:44:52.000000000","message":"Patch Set 2:\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"97db407b4f500a60cd351f2739265edaf35a3d13","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-01-28 17:07:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"416a073562463ebd62549876e83bfee650c8eea7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 17:07:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a691683c4215efae63a6fb3bf5e910b53227abb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 17:16:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7923c7848d514344977f4f4156d311eb : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":2},{"id":"c6cd96a36011294d32a035e3be961576d6aa4095","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 17:16:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"13a9a89f5f03bbab7fc59ea375ea331711c244d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 17:18:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/58656cfb74ef4f339c0536574ee58ea5 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"09f8b8b4b4b65e440af6c11fa940bdea83340370","revisions":{"b24ccda6a920b747566845affbbcddd9a1549f78":{"kind":"REWORK","_number":1,"created":"2019-12-13 16:22:58.000000000","uploader":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"ref":"refs/changes/51/698951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/51/698951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/51/698951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/51/698951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/51/698951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/51/698951/1"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2019-12-13 16:21:39.000000000","tz":-300},"committer":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2019-12-13 16:21:39.000000000","tz":-300},"subject":"Repropose Expiring Group Membership for Ussuri","message":"Repropose Expiring Group Membership for Ussuri\n\nAdd federated users to the groups that they receive from the mapping rules.\nThis membership is only carried by the token and not persisted in the\ndatabase. The membership expires, but can be renewed when the user\nauthenticates with the same group.\n\nPreviously approved for Train, fell into backlog, reproposing for Ussuri.\n\nChange-Id: Ie133c14ffba5e4189265920759bfb5e1391f1189\nPartial-Bug: 1809116\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b24ccda6a920b747566845affbbcddd9a1549f78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b24ccda6a920b747566845affbbcddd9a1549f78"}]},"branch":"refs/heads/master"},"09f8b8b4b4b65e440af6c11fa940bdea83340370":{"kind":"REWORK","_number":2,"created":"2019-12-17 15:09:39.000000000","uploader":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"ref":"refs/changes/51/698951/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/51/698951/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/51/698951/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/51/698951/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/51/698951/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/51/698951/2"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2019-12-13 16:21:39.000000000","tz":-300},"committer":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2019-12-17 15:09:31.000000000","tz":-300},"subject":"Repropose Expiring Group Membership for Ussuri","message":"Repropose Expiring Group Membership for Ussuri\n\nAdd federated users to the groups that they receive from the mapping rules.\nThis membership is only carried by the token and not persisted in the\ndatabase. The membership expires, but can be renewed when the user\nauthenticates with the same group.\n\nPreviously approved for Train, fell into backlog, reproposing for Ussuri.\n\nChange-Id: Ie133c14ffba5e4189265920759bfb5e1391f1189\nPartial-Bug: 1809116\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/09f8b8b4b4b65e440af6c11fa940bdea83340370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/09f8b8b4b4b65e440af6c11fa940bdea83340370"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
