)]}'
{"id":"openstack%2Fkeystone-specs~699433","triplet_id":"openstack%2Fkeystone-specs~master~I01b7f61bf6d07d00bec6802a56fcd977df1eb5ec","project":"openstack/keystone-specs","branch":"master","topic":"drop-py27-support","hashtags":[],"change_id":"I01b7f61bf6d07d00bec6802a56fcd977df1eb5ec","subject":"[ussuri][goal] Drop python 2.7 support","status":"ABANDONED","created":"2019-12-17 15:11:27.000000000","updated":"2019-12-19 09:20:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b959f479d00158cfe762422991bf420749d3010a","_number":699433,"virtual_id_number":699433,"owner":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-17 15:19:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"all":[{"value":-1,"date":"2019-12-17 15:52:42.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 15:19:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e42e6b03bd6fc3d859f061d5a66a006d9ac2b22","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-12-17 15:11:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33f9be31518867014658ae21a5ee5e9cfeeedf26","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-12-17 15:12:24.000000000","message":"Topic changed from 698279 to drop-py27-support","accounts_in_message":[],"_revision_number":1},{"id":"f59deb74a1345606ef025d8d51cc80113992ad8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 15:19:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93a2306227cf48289310bf35bad84e01 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23ff5bafc8014335ae0cce39171b0db4 : SUCCESS in 4m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4c266818d77494a94bddcd6d15017e1 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ba8008f967e43adabcc9741ae5532cc : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"59b22d5d186d611da25d8dccd210354b529eb5d5","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-12-17 15:52:42.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b959f479d00158cfe762422991bf420749d3010a","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-12-19 09:20:56.000000000","message":"Abandoned\n\nhttps://review.opendev.org/#/c/698983 gate is passed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f5319cef5bed30f3ec1804c5d470614d36102aa5","revisions":{"f5319cef5bed30f3ec1804c5d470614d36102aa5":{"kind":"REWORK","_number":1,"created":"2019-12-17 15:11:27.000000000","uploader":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"ref":"refs/changes/33/699433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/33/699433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/33/699433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/33/699433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/33/699433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/33/699433/1"}}},"commit":{"parents":[{"commit":"b90f7c87021fb3eb9cb093a5af9ef9670e0af347","subject":"Set up for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b90f7c87021fb3eb9cb093a5af9ef9670e0af347"}]}],"author":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-17 15:10:27.000000000","tz":330},"committer":{"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","date":"2019-12-17 15:10:27.000000000","tz":330},"subject":"[ussuri][goal] Drop python 2.7 support","message":"[ussuri][goal] Drop python 2.7 support\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nspecs repo either has py27 job or requirement or tox env.\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: I01b7f61bf6d07d00bec6802a56fcd977df1eb5ec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f5319cef5bed30f3ec1804c5d470614d36102aa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f5319cef5bed30f3ec1804c5d470614d36102aa5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
