)]}'
{"id":"openstack%2Fkeystone-specs~748042","triplet_id":"openstack%2Fkeystone-specs~master~I84426119a60c595eaf462662c6a24a990914c516","project":"openstack/keystone-specs","branch":"master","topic":"bp/versioning-for-attribute-mapping-schema","attention_set":{},"removed_from_attention_set":{"16465":{"account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"last_update":"2023-11-20 21:43:05.000000000","reason":"\u003cGERRIT_ACCOUNT_16465\u003e replied on the change","reason_account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},"8482":{"account":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"last_update":"2023-12-06 15:05:36.000000000","reason":"Change was submitted"},"28356":{"account":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"last_update":"2023-12-06 15:05:36.000000000","reason":"Change was submitted"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2023-12-06 15:05:36.000000000","reason":"Change was submitted"},"13478":{"account":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"last_update":"2023-12-06 15:05:36.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2022-07-01 15:14:54.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2023-11-21 07:01:50.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"I84426119a60c595eaf462662c6a24a990914c516","subject":"Add schema version and support to \"domain\" attribute in mapping rules","status":"MERGED","created":"2020-08-25 22:31:52.000000000","updated":"2023-12-12 20:08:10.000000000","submitted":"2023-12-06 15:05:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":77,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"748042-bp/versioning-for-attribute-mapping-schema","meta_rev_id":"bb8c290d4ae10dbd84222e3b71a4ee6b63119427","_number":748042,"virtual_id_number":748042,"owner":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-06 15:05:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2023-11-20 21:43:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2023-12-06 14:55:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2023-11-21 07:01:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2023-12-06 14:55:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-28 20:10:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-10 16:17:31.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-11-16 09:33:07.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2020-11-22 22:52:20.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-12-08 21:27:52.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2022-02-22 15:21:00.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-02-22 15:21:00.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-05-13 04:38:17.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REMOVED"},{"updated":"2023-10-10 14:48:25.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"CC"},{"updated":"2023-10-10 14:48:25.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2023-11-18 07:46:20.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2023-11-29 15:27:54.000000000","updated_by":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"reviewer":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"state":"CC"}],"messages":[{"id":"ddad39a065d333837a763ce774ac287a3a8776b2","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-08-25 22:31:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77e4d6da98da758fa2e1fb111f49a8269d0b7470","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 22:39:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/049f4b229a3f4a648213a2e22717d780 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30f684cade974902835e8bb8cf5c3edc : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2907fc274cb7e8517ae2ca4504a08450d27307f0","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-08-25 22:42:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7b3ab93fb22d2a3948265c276461787ff733a8a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 22:48:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fee4fdbb799048e1a72678382bc7b4ac : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/232c03b77dfa4316a75c0f27a07590d7 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":2},{"id":"30263297bfc2e59ed58d27783d8cd0740534a38e","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-08-26 20:24:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"25b5b36babd6434477f3a4b81282b713778503b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 20:36:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adfb8a7196de43b89549bb4b8a6373c0 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6b9e8acfb214a70908f48da363e30bd : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":3},{"id":"52b71fe0336ef298f27b4498537e58857b893379","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-08-28 20:00:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2254ae951baf6b1bc15e23d280e2899abca038d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 20:10:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f57fc968e245476faa6d152c4aa45e64 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e20f26012f2e4beabb9eadb9d389135f : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":4},{"id":"cb59a97662127d295f9478d5aac607c1e2fec08a","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-09-08 14:38:21.000000000","message":"Patch Set 4: Code-Review+1\n\nHi Rafael, this looks pretty good and agrees with the general direction of the project. Thanks for going into detail and incorporating the attribute into the schema of the mapping as we talked about in the meeting a few weeks ago. I will upgrade to a +2 after I have thought about federation plans a bit more, but this is a great target for Wallaby.","accounts_in_message":[],"_revision_number":4},{"id":"cf4cebf2cf42ee0bba2ae51ee0f2df35cd73a8ea","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-11-10 16:17:31.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4fc1ffba4edc6c7bdac295a9f468a4a35ee52cd6","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-11-16 09:33:07.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"555eace5fd1557f393abdb68777ed2a3e33da9f2","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-11-22 22:52:20.000000000","message":"Patch Set 4: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":4},{"id":"219730383ba726126a1c0e833038d4211f8b78bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-11-25 14:14:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b3c825fb20166661bee5cefd1a2f7a13f21bb25c","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-11-25 14:15:09.000000000","message":"Patch Set 4:\n\n(7 comments)\n\nThanks Collen for your reviews!\nI have replied them, and applied some adjusts to the spec.","accounts_in_message":[],"_revision_number":4},{"id":"6ae9568c0f966bbb7920d33e0f11c4a4ab13ebcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 14:30:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a220787ec9a48c09628c24daed36570 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae4bcbe686c347c598bf73b3cf5676df : FAILURE in 3m 19s","accounts_in_message":[],"_revision_number":5},{"id":"f97b0010ef899608a7a9b17f7137e2e65a78ad93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-11-25 14:43:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"59a08a9ab048ee3c91971ac7ea851a979dc73194","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 15:03:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/249bf8b1c1334eff914b366a5cb1c27e : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b872ae959eb641b983f9256067752e7b : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":6},{"id":"2891aa75259ecefffd335f0dac124cb253f63b86","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-11-25 18:29:36.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b49669460d339286f6e4422b48863a124ea5b8b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-11-25 19:02:04.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"269b46ad1d15f497aea5c6592f978ecd63bfed55","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-11-25 19:02:48.000000000","message":"Patch Set 7:\n\n(7 comments)","accounts_in_message":[],"_revision_number":7},{"id":"7d1ea80786a8fedfd55a22faf71848821447485a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 19:12:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740c367871644e7dbcf297c10c81b881 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a92ff5160f8c42aa937300ae628690fa : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":7},{"id":"554ba332cd17ff1a4a20d83930d0f23df183f7cb","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-11-30 19:58:50.000000000","message":"Patch Set 7: Code-Review+1\n\n(2 comments)\n\nI still would hope to see the JSON API documented in the spec, but I don\u0027t want to continue to block the spec. If other reviewers are happy with this and with the proposed behavior of how the domain is interpreted is fine with them then it is fine with me.","accounts_in_message":[],"_revision_number":7},{"id":"c5a4bb113348b9a64a4c7f270f11baa3069cf211","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-12-01 11:52:31.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7898fb2edaf82023dad7a00d7518dd590ccb415e","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-12-01 11:53:19.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"4ca7228dadfacfc92b8f623ee40399c73f7d604d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 12:03:28.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4df6238a8b5b4dff96e7c409ea701b22 : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f3f4e11b3e8437c8c05b50567a97b9b : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":8},{"id":"a22870a619d3e28df4a86e15d5255e69ba3710d6","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-12-08 21:27:52.000000000","message":"Patch Set 8: Code-Review+1\n\nLooks ok to me, would like someone with more knowledge of federation to look over this as well.","accounts_in_message":[],"_revision_number":8},{"id":"ff0af9fe1585fb25884c2c9bdcfda35bba9bdec3","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-12-21 11:16:24.000000000","message":"Patch Set 8:\n\nIs there something that we need to address here, or is this one ready to be merged? I would like to use some free time in the holidays to work on the Patches that introduce this spec.","accounts_in_message":[],"_revision_number":8},{"id":"4fa28250d44885e9152421a19454c83e8e32ed6a","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2021-02-17 11:31:15.000000000","message":"Patch Set 8:\n\nTwo months have gone by. Are we still missing something here?","accounts_in_message":[],"_revision_number":8},{"id":"14421c0cea10b3d4df4b8a33e7f5705d354a092e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2022-02-23 13:02:38.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8f94a5f44a35b2c6f13ee9fd0c4f89b41c4437ba","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2022-02-23 13:04:58.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5a60201fd3a0b2cd0d9b108cf1ef4e426fe65cc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 13:09:34.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38dace4b27334a7f93510be7e21cdee0 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62e0dbcd02024b5180f92b430999ef13 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":9},{"id":"94a9f8cc4dc0e576f4ecb5102827df3ded1354bc","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-05-13 04:38:17.000000000","message":"Removed reviewer Gage Hugo.","accounts_in_message":[],"_revision_number":9},{"id":"d5ec4ebfe15536d88dbf76c4dc4f888a1071b893","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2022-06-16 15:58:17.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c0728ad41d8a07119bb3acc97243af38fb928bd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2022-06-24 13:38:09.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"db34cdd62c4c5b2dd5d3f9664b2dd4dfbb216ddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 13:44:50.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51c2cfc00c904ed4b02457e252bcb4e7 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9728781aceb1492bb4845572cc3a5115 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":10},{"id":"1aeb66efa9162b23b5839038d6240172b92bba0f","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-07-01 14:48:30.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f605056accac694512aacc2f588fdad6bfacf008","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-07-01 14:59:36.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1057464e1999def190b9c465e36c3f4de4aa6081","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-07-01 15:14:54.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"999b5691cf66048ffa34590588623dd6f611d146","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2022-07-01 19:57:15.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"ed87cfcd3a643c71a8b0c90b1261e2acb95db267","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2022-07-01 19:58:32.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"b7481eb433a5a698ff85958e1970707ea2f583d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 20:02:25.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1caaac5ff60a46e3836d3ac3f2890380 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08f51794d53e413b93871eeca0ecfb4b : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":11},{"id":"91f44468a354d60f08201434dc6a838493f1f9f1","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-07-01 22:11:56.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"6166fad6daf2890c335ccac74a1a1b6e8171b7e7","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-07-29 14:33:27.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5f6050386634e1f7f6b068d8197d537ba85f57c4","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2022-08-19 01:41:01.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"1948874a02dcc9ced919a1220b5645b8846b2c50","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2022-11-30 11:48:37.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e4ae20be6f6b205c27774cd53abf728a1cee369c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2022-12-06 15:02:37.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"bcfe90c8278a567486a8cb98dcd8d859be5330e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-06 15:09:37.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/663222bb297642c68b9f85556c914cb4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/992192c77a2f4c568bd715fa11c067f2 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e0ec49609914c09b6c874977c03139d : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":12},{"id":"2e57ae588aec0d00fffdad348951c6d0509a5960","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-10-10 14:48:25.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"473787363c28a016ec7c850f8df7fdcbca7605f4","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-10-13 18:42:24.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"71020340aacbea108e016d1a313cb9f3039223d7","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-10-13 18:49:06.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"fe6fdbdcca2c4a085f14063b153b4da6602412a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-10-13 18:56:38.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"749b77a78a4d8a2922df6ed48dc3b8abb282502d","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-10-13 18:57:04.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"67fc0b9fcdbd9ca7e553f69de71aa9d34e07adb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-10-13 18:58:14.000000000","message":"Uploaded patch set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"25783372692854df37d05f1dff88b1d400eb3ccf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 19:02:02.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/149b1bf029214ae18ec7d1518b95f867\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd63af0706c4442e9ff88f15f4eb0daa : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97846987a8814a8bb75adcc936ef7bb6 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":14},{"id":"9716f0924b51c89c8bc240a1d7c71592f672ae01","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-10-23 13:46:51.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"2466deadeac026d4fbbde81742c48d69043bd847","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-11-18 07:46:20.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"3edfad0572885fa8f8dac6bcc4e57b7bd229a8cf","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-11-18 11:11:17.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"f1467351f9e6c6f556f07e0df4f85ccedff8ec0e","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2023-11-18 21:18:04.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"4102017d3c7325d324965f1d37f969e9e5537fc2","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-11-18 23:28:08.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"57ec355e3aaf6732a8db3dccb4f6068d2783da4b","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2023-11-19 02:15:53.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"53b42e30cea4fbed5a05b9e7e68b9afc2c7cba75","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-11-19 23:50:35.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"e86bfe9bac4db861d70d27141bfb5e6bb4c8bd50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-11-19 23:50:36.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"d24b23fdeb84aec09d24598b76db1e9e992e3ae4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-19 23:54:42.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8f3e617c1a94b089be7f3f625735f20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/724a7d2410fc4b03bafd5f150458f180 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/483c311493d2476f8d4f7c503e3474fb : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":15},{"id":"6e4580a55767220ac73d22f5868933030d0581f3","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-11-20 05:28:21.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"57853b7650e970bc16e6f39f107770e032f50a81","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2023-11-20 19:14:21.000000000","message":"Patch Set 15: Code-Review+1\n\n(8 comments)","accounts_in_message":[],"_revision_number":15},{"id":"8847b58260f216c922d8a57ca10f7526851ea480","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2023-11-20 19:15:20.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"e8c941ea4043108ac6190cf6ba4c26d4571b2309","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-11-20 20:21:51.000000000","message":"Patch Set 15:\n\n(8 comments)","accounts_in_message":[],"_revision_number":15},{"id":"e647d4d932454cb6e470f0278c17044f98e48f3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-11-20 20:22:24.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"7f432005d16405bc49e9ce880dfa4f9c45b9707e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 20:31:12.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98a5475caed24b00bf549dff8df76d6a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f847452d0b954deb8c34472a8f29975a : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6133effcf9b4511ac3ee29b4164485c : FAILURE in 2m 47s","accounts_in_message":[],"_revision_number":16},{"id":"8b1b1a6ced5818b24d04a3d64d2ece2aa9e92425","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-11-20 20:41:01.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"a2cfdd0306e495e103f0766c30ac9942be58b0c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 20:49:12.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bbfcf1d15f34fa3a083827f0987af5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7918e9dce68840f99c68b29733fd75a6 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/007fccb3e5444ea1b800be301da35178 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":17},{"id":"2047469498072f83ac7cd1a91d07c5f3ede45944","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2023-11-20 21:43:05.000000000","message":"Patch Set 17: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"e2db6e1a38ae2b8cc4c2bee2b4245de0855c14d4","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-11-21 07:01:50.000000000","message":"Patch Set 17: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"486ef69aabfe4900d0b59f513a3ba2fc2dcff971","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2023-11-29 15:27:54.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"7da2c48b7dc3f5c7266217ebd312f28dd2289835","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-11-29 16:28:13.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"94ed7967103a2c9f0751d3e8392fdd9163497155","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2023-11-29 16:40:32.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"e06ebf9a3629f8da0f880f105d7d5a687bc2abcd","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-11-30 17:40:13.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"834d5e8e693bd0476dc1bbaf8283d96569e2298b","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-12-06 14:55:37.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"22421795ff88d9cd01d8265834d1da2d3fcfcb7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 14:56:08.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"afd93bc0af7134f64524217a1797f9ec533def04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 15:05:34.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9130211fe2a743ddacfc2c08185bf2f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ebe11f95bc547e18398c2d2e8558c98 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b34cb60ff39485eb48dd5d9c3acdfb6 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":17},{"id":"852018ec935d7fd33fd576d5e29befe2c0d18a6c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 15:05:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"fae8e23ab05604605378ac4a57d6ddda49c97b75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-06 15:06:42.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6519e848722b42e99ea40262ef7140d6\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/0af28d33d6ac47f99f24178d1389a691 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":17},{"id":"bb8c290d4ae10dbd84222e3b71a4ee6b63119427","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2023-12-12 20:08:10.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"2170b9176d6d8edaecb2bfee998f9a2d082b180c","revisions":{"5698cc42670f974d8e47035e8f2729cf329de894":{"kind":"REWORK","_number":1,"created":"2020-08-25 22:31:52.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/42/748042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/42/748042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/1"}}},"commit":{"parents":[{"commit":"70f9caa74060171df8eb63b2688f88b7fce17184","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/70f9caa74060171df8eb63b2688f88b7fce17184"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:22:07.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:31:50.000000000","tz":-180},"subject":"Keystone to honor the \"domain\" attribute mapping rules","message":"Keystone to honor the \"domain\" attribute mapping rules\n\nAs discussed in the Keystone meeting [1], I am proposing this spec, so we can\nreview and discuss the implementation proposed in [2].\n\n[1] http://eavesdrop.openstack.org/meetings/keystone/2020/keystone.2020-08-25-16.59.log.txt\n[2] https://review.opendev.org/#/c/739966/\n\nChange-Id: I84426119a60c595eaf462662c6a24a990914c516\nImplements: blueprint versioning-for-attribute-mapping-schema\nRFE: https://bugs.launchpad.net/keystone/+bug/1887515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/5698cc42670f974d8e47035e8f2729cf329de894"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/5698cc42670f974d8e47035e8f2729cf329de894"}]},"branch":"refs/heads/master"},"b5b232ec2bd610a9bf5aefd62e453d983009a15d":{"kind":"REWORK","_number":2,"created":"2020-08-25 22:42:40.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/42/748042/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/42/748042/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/2"}}},"commit":{"parents":[{"commit":"70f9caa74060171df8eb63b2688f88b7fce17184","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/70f9caa74060171df8eb63b2688f88b7fce17184"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:22:07.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:42:28.000000000","tz":-180},"subject":"Keystone to honor the \"domain\" attribute mapping rules","message":"Keystone to honor the \"domain\" attribute mapping rules\n\nAs discussed in the Keystone meeting [1], I am proposing this spec, so we can\nreview and discuss the implementation proposed in [2].\n\n[1] http://eavesdrop.openstack.org/meetings/keystone/2020/keystone.2020-08-25-16.59.log.txt\n[2] https://review.opendev.org/#/c/739966/\n\nChange-Id: I84426119a60c595eaf462662c6a24a990914c516\nImplements: blueprint versioning-for-attribute-mapping-schema\nRFE: https://bugs.launchpad.net/keystone/+bug/1887515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b5b232ec2bd610a9bf5aefd62e453d983009a15d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b5b232ec2bd610a9bf5aefd62e453d983009a15d"}]},"branch":"refs/heads/master"},"25ce7adb9c7c8db8d4d97c1725ec5c4a46cfba18":{"kind":"REWORK","_number":3,"created":"2020-08-26 20:24:32.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/42/748042/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/42/748042/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/3"}}},"commit":{"parents":[{"commit":"70f9caa74060171df8eb63b2688f88b7fce17184","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/70f9caa74060171df8eb63b2688f88b7fce17184"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:22:07.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-26 20:23:59.000000000","tz":-180},"subject":"Keystone to honor the \"domain\" attribute mapping rules","message":"Keystone to honor the \"domain\" attribute mapping rules\n\nAs discussed in the Keystone meeting [1], I am proposing this spec, so we can\nreview and discuss the implementation proposed in [2].\n\n[1] http://eavesdrop.openstack.org/meetings/keystone/2020/keystone.2020-08-25-16.59.log.txt\n[2] https://review.opendev.org/#/c/739966/\n\nChange-Id: I84426119a60c595eaf462662c6a24a990914c516\nImplements: blueprint versioning-for-attribute-mapping-schema\nRFE: https://bugs.launchpad.net/keystone/+bug/1887515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/25ce7adb9c7c8db8d4d97c1725ec5c4a46cfba18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/25ce7adb9c7c8db8d4d97c1725ec5c4a46cfba18"}]},"branch":"refs/heads/master"},"c0b1e99db2d12a5c900a6b39fdc180783bd0c0de":{"kind":"REWORK","_number":4,"created":"2020-08-28 20:00:05.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/42/748042/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/42/748042/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/4"}}},"commit":{"parents":[{"commit":"70f9caa74060171df8eb63b2688f88b7fce17184","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/70f9caa74060171df8eb63b2688f88b7fce17184"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:22:07.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-28 19:59:25.000000000","tz":-180},"subject":"Keystone to honor the \"domain\" attribute mapping rules","message":"Keystone to honor the \"domain\" attribute mapping rules\n\nAs discussed in the Keystone meeting [1], I am proposing this spec, so we can\nreview and discuss the implementation proposed in [2].\n\n[1] http://eavesdrop.openstack.org/meetings/keystone/2020/keystone.2020-08-25-16.59.log.txt\n[2] https://review.opendev.org/#/c/739966/\n\nChange-Id: I84426119a60c595eaf462662c6a24a990914c516\nImplements: blueprint versioning-for-attribute-mapping-schema\nRFE: https://bugs.launchpad.net/keystone/+bug/1887515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/c0b1e99db2d12a5c900a6b39fdc180783bd0c0de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/c0b1e99db2d12a5c900a6b39fdc180783bd0c0de"}]},"branch":"refs/heads/master"},"1591ede8c9bf798a91ee05a88a2e19addbeb7b44":{"kind":"REWORK","_number":5,"created":"2020-11-25 14:14:14.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/42/748042/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/42/748042/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/5"}}},"commit":{"parents":[{"commit":"70f9caa74060171df8eb63b2688f88b7fce17184","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/70f9caa74060171df8eb63b2688f88b7fce17184"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:22:07.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-11-25 14:13:49.000000000","tz":-180},"subject":"Keystone to honor the \"domain\" attribute mapping rules","message":"Keystone to honor the \"domain\" attribute mapping rules\n\nAs discussed in the Keystone meeting [1], I am proposing this spec, so we can\nreview and discuss the implementation proposed in [2].\n\n[1] http://eavesdrop.openstack.org/meetings/keystone/2020/keystone.2020-08-25-16.59.log.txt\n[2] https://review.opendev.org/#/c/739966/\n\nChange-Id: I84426119a60c595eaf462662c6a24a990914c516\nImplements: blueprint versioning-for-attribute-mapping-schema\nRFE: https://bugs.launchpad.net/keystone/+bug/1887515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/1591ede8c9bf798a91ee05a88a2e19addbeb7b44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/1591ede8c9bf798a91ee05a88a2e19addbeb7b44"}]},"branch":"refs/heads/master"},"80d5f75206ea9e966697846cb7ed13a621d1a362":{"kind":"REWORK","_number":6,"created":"2020-11-25 14:43:18.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/42/748042/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/42/748042/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/6"}}},"commit":{"parents":[{"commit":"70f9caa74060171df8eb63b2688f88b7fce17184","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/70f9caa74060171df8eb63b2688f88b7fce17184"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:22:07.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-11-25 14:43:05.000000000","tz":-180},"subject":"Keystone to honor the \"domain\" attribute mapping rules","message":"Keystone to honor the \"domain\" attribute mapping rules\n\nAs discussed in the Keystone meeting [1], I am proposing this spec, so we can\nreview and discuss the implementation proposed in [2].\n\n[1] http://eavesdrop.openstack.org/meetings/keystone/2020/keystone.2020-08-25-16.59.log.txt\n[2] https://review.opendev.org/#/c/739966/\n\nChange-Id: I84426119a60c595eaf462662c6a24a990914c516\nImplements: blueprint versioning-for-attribute-mapping-schema\nRFE: https://bugs.launchpad.net/keystone/+bug/1887515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/80d5f75206ea9e966697846cb7ed13a621d1a362"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/80d5f75206ea9e966697846cb7ed13a621d1a362"}]},"branch":"refs/heads/master"},"66439ee390ea3811354059ea5fc4078214788fb4":{"kind":"REWORK","_number":7,"created":"2020-11-25 19:02:04.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/42/748042/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/42/748042/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/7"}}},"commit":{"parents":[{"commit":"70f9caa74060171df8eb63b2688f88b7fce17184","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/70f9caa74060171df8eb63b2688f88b7fce17184"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:22:07.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-11-25 19:01:44.000000000","tz":-180},"subject":"Keystone to honor the \"domain\" attribute mapping rules","message":"Keystone to honor the \"domain\" attribute mapping rules\n\nAs discussed in the Keystone meeting [1], I am proposing this spec, so we can\nreview and discuss the implementation proposed in [2].\n\n[1] http://eavesdrop.openstack.org/meetings/keystone/2020/keystone.2020-08-25-16.59.log.txt\n[2] https://review.opendev.org/#/c/739966/\n\nChange-Id: I84426119a60c595eaf462662c6a24a990914c516\nImplements: blueprint versioning-for-attribute-mapping-schema\nRFE: https://bugs.launchpad.net/keystone/+bug/1887515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/66439ee390ea3811354059ea5fc4078214788fb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/66439ee390ea3811354059ea5fc4078214788fb4"}]},"branch":"refs/heads/master"},"0b825da9e715676a1172578b08b80dcfde0ec875":{"kind":"REWORK","_number":8,"created":"2020-12-01 11:52:31.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/42/748042/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/42/748042/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/8"}}},"commit":{"parents":[{"commit":"70f9caa74060171df8eb63b2688f88b7fce17184","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/70f9caa74060171df8eb63b2688f88b7fce17184"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:22:07.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-12-01 11:52:19.000000000","tz":-180},"subject":"Keystone to honor the \"domain\" attribute mapping rules","message":"Keystone to honor the \"domain\" attribute mapping rules\n\nAs discussed in the Keystone meeting [1], I am proposing this spec, so we can\nreview and discuss the implementation proposed in [2].\n\n[1] http://eavesdrop.openstack.org/meetings/keystone/2020/keystone.2020-08-25-16.59.log.txt\n[2] https://review.opendev.org/#/c/739966/\n\nChange-Id: I84426119a60c595eaf462662c6a24a990914c516\nImplements: blueprint versioning-for-attribute-mapping-schema\nRFE: https://bugs.launchpad.net/keystone/+bug/1887515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/0b825da9e715676a1172578b08b80dcfde0ec875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/0b825da9e715676a1172578b08b80dcfde0ec875"}]},"branch":"refs/heads/master"},"e68fc57d8ef94b764a699253570e409326ed8522":{"kind":"REWORK","_number":9,"created":"2022-02-23 13:02:38.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/42/748042/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/42/748042/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/9"}}},"commit":{"parents":[{"commit":"8145886d241a48e9dd418c4b56a3a200a0ffe736","subject":"OAuth2.0 Client Credentials Grant Flow Support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/8145886d241a48e9dd418c4b56a3a200a0ffe736"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:22:07.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2022-02-23 13:02:23.000000000","tz":-180},"subject":"Keystone to honor the \"domain\" attribute mapping rules","message":"Keystone to honor the \"domain\" attribute mapping rules\n\nAs discussed in the Keystone meeting [1], I am proposing this spec, so we can\nreview and discuss the implementation proposed in [2].\n\n[1] http://eavesdrop.openstack.org/meetings/keystone/2020/keystone.2020-08-25-16.59.log.txt\n[2] https://review.opendev.org/#/c/739966/\n\nChange-Id: I84426119a60c595eaf462662c6a24a990914c516\nImplements: blueprint versioning-for-attribute-mapping-schema\nRFE: https://bugs.launchpad.net/keystone/+bug/1887515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/e68fc57d8ef94b764a699253570e409326ed8522"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/e68fc57d8ef94b764a699253570e409326ed8522"}]},"branch":"refs/heads/master"},"9d1efea8013ba1c240af7de2e25ccfcdffd0549c":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2022-06-24 13:38:09.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/42/748042/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/42/748042/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/10"}}},"commit":{"parents":[{"commit":"6df4f46055c64e877ce555123b5be7ab395429cc","subject":"Disable auto-discovery for setuptools and update python testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6df4f46055c64e877ce555123b5be7ab395429cc"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:22:07.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2022-06-24 13:34:20.000000000","tz":-180},"subject":"Keystone to honor the \"domain\" attribute mapping rules","message":"Keystone to honor the \"domain\" attribute mapping rules\n\nAs discussed in the Keystone meeting [1], I am proposing this spec, so we can\nreview and discuss the implementation proposed in [2].\n\n[1] http://eavesdrop.openstack.org/meetings/keystone/2020/keystone.2020-08-25-16.59.log.txt\n[2] https://review.opendev.org/#/c/739966/\n\nChange-Id: I84426119a60c595eaf462662c6a24a990914c516\nImplements: blueprint versioning-for-attribute-mapping-schema\nRFE: https://bugs.launchpad.net/keystone/+bug/1887515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/9d1efea8013ba1c240af7de2e25ccfcdffd0549c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/9d1efea8013ba1c240af7de2e25ccfcdffd0549c"}]},"branch":"refs/heads/master"},"5e224b91a2307945bb03491d58eec9dbdae929fb":{"kind":"REWORK","_number":11,"created":"2022-07-01 19:57:15.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/42/748042/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/42/748042/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/11"}}},"commit":{"parents":[{"commit":"6df4f46055c64e877ce555123b5be7ab395429cc","subject":"Disable auto-discovery for setuptools and update python testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6df4f46055c64e877ce555123b5be7ab395429cc"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:22:07.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2022-07-01 19:56:58.000000000","tz":-180},"subject":"Add schema version and add support to \"domain\" attribute in mapping rules","message":"Add schema version and add support to \"domain\" attribute in mapping rules\n\nAs discussed in the Keystone meeting [1], I am proposing this spec, so we can\nreview and discuss the implementation proposed in [2].\n\n[1] http://eavesdrop.openstack.org/meetings/keystone/2020/keystone.2020-08-25-16.59.log.txt\n[2] https://review.opendev.org/#/c/739966/\n\nChange-Id: I84426119a60c595eaf462662c6a24a990914c516\nImplements: blueprint versioning-for-attribute-mapping-schema\nRFE: https://bugs.launchpad.net/keystone/+bug/1887515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/5e224b91a2307945bb03491d58eec9dbdae929fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/5e224b91a2307945bb03491d58eec9dbdae929fb"}]},"branch":"refs/heads/master"},"87cb1e6bde224371b1feba406dd383e3e0a97124":{"kind":"REWORK","_number":12,"created":"2022-12-06 15:02:37.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/42/748042/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/42/748042/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/12"}}},"commit":{"parents":[{"commit":"75b4fb25c5b6e2c6ef4b0157c98eeb17c777c054","subject":"Describe the need for a default service role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/75b4fb25c5b6e2c6ef4b0157c98eeb17c777c054"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:22:07.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2022-12-06 15:02:24.000000000","tz":-180},"subject":"Add schema version and add support to \"domain\" attribute in mapping rules","message":"Add schema version and add support to \"domain\" attribute in mapping rules\n\nAs discussed in the Keystone meeting [1], I am proposing this spec, so we can\nreview and discuss the implementation proposed in [2].\n\n[1] http://eavesdrop.openstack.org/meetings/keystone/2020/keystone.2020-08-25-16.59.log.txt\n[2] https://review.opendev.org/#/c/739966/\n\nChange-Id: I84426119a60c595eaf462662c6a24a990914c516\nImplements: blueprint versioning-for-attribute-mapping-schema\nRFE: https://bugs.launchpad.net/keystone/+bug/1887515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/87cb1e6bde224371b1feba406dd383e3e0a97124"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/87cb1e6bde224371b1feba406dd383e3e0a97124"}]},"branch":"refs/heads/master"},"310f351a8103a1ef8c363a3abe95f658aa60c4b4":{"kind":"REWORK","_number":13,"created":"2023-10-13 18:56:38.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/42/748042/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/42/748042/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/13"}}},"commit":{"parents":[{"commit":"330200b4d8c6558a8ea2964d521b5d2f659a48c0","subject":"Merge \"External OAuth2.0 Authorization Server Support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/330200b4d8c6558a8ea2964d521b5d2f659a48c0"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:22:07.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-10-13 18:56:27.000000000","tz":-180},"subject":"Add schema version and add support to \"domain\" attribute in mapping rules","message":"Add schema version and add support to \"domain\" attribute in mapping rules\n\nAs discussed in the Keystone meeting [1], I am proposing this spec, so we can\nreview and discuss the implementation proposed in [2].\n\n[1] http://eavesdrop.openstack.org/meetings/keystone/2020/keystone.2020-08-25-16.59.log.txt\n[2] https://review.opendev.org/#/c/739966/\n\nChange-Id: I84426119a60c595eaf462662c6a24a990914c516\nImplements: blueprint versioning-for-attribute-mapping-schema\nRFE: https://bugs.launchpad.net/keystone/+bug/1887515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/310f351a8103a1ef8c363a3abe95f658aa60c4b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/310f351a8103a1ef8c363a3abe95f658aa60c4b4"}]},"branch":"refs/heads/master"},"8c21bee4aaa72aef90e0dd7ca573678e6304895e":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2023-10-13 18:58:14.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/42/748042/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/42/748042/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/14"}}},"commit":{"parents":[{"commit":"330200b4d8c6558a8ea2964d521b5d2f659a48c0","subject":"Merge \"External OAuth2.0 Authorization Server Support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/330200b4d8c6558a8ea2964d521b5d2f659a48c0"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:22:07.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-10-13 18:57:41.000000000","tz":-180},"subject":"Add schema version and support to \"domain\" attribute in mapping rules","message":"Add schema version and support to \"domain\" attribute in mapping rules\n\nAs discussed in the Keystone meeting [1], I am proposing this spec, so\nwe can review and discuss the implementation proposed in [2].\n\n[1] http://eavesdrop.openstack.org/meetings/keystone/2020/keystone.2020-08-25-16.59.log.txt\n[2] https://review.opendev.org/#/c/739966/\n\nChange-Id: I84426119a60c595eaf462662c6a24a990914c516\nImplements: blueprint versioning-for-attribute-mapping-schema\nRFE: https://bugs.launchpad.net/keystone/+bug/1887515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/8c21bee4aaa72aef90e0dd7ca573678e6304895e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/8c21bee4aaa72aef90e0dd7ca573678e6304895e"}]},"branch":"refs/heads/master"},"239c75547354f7bd3566248c9c591727ab924970":{"kind":"REWORK","_number":15,"created":"2023-11-19 23:50:36.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/42/748042/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/42/748042/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/15"}}},"commit":{"parents":[{"commit":"330200b4d8c6558a8ea2964d521b5d2f659a48c0","subject":"Merge \"External OAuth2.0 Authorization Server Support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/330200b4d8c6558a8ea2964d521b5d2f659a48c0"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:22:07.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-11-19 23:50:24.000000000","tz":-180},"subject":"Add schema version and support to \"domain\" attribute in mapping rules","message":"Add schema version and support to \"domain\" attribute in mapping rules\n\nAs discussed in the Keystone meeting [1], I am proposing this spec, so\nwe can review and discuss the implementation proposed in [2].\n\n[1] http://eavesdrop.openstack.org/meetings/keystone/2020/keystone.2020-08-25-16.59.log.txt\n[2] https://review.opendev.org/#/c/739966/\n\nChange-Id: I84426119a60c595eaf462662c6a24a990914c516\nImplements: blueprint versioning-for-attribute-mapping-schema\nRFE: https://bugs.launchpad.net/keystone/+bug/1887515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/239c75547354f7bd3566248c9c591727ab924970"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/239c75547354f7bd3566248c9c591727ab924970"}]},"branch":"refs/heads/master"},"62bcfa6f3fa28ed4e058c372db7646ca0110b409":{"kind":"REWORK","_number":16,"created":"2023-11-20 20:22:24.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/42/748042/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/42/748042/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/16"}}},"commit":{"parents":[{"commit":"330200b4d8c6558a8ea2964d521b5d2f659a48c0","subject":"Merge \"External OAuth2.0 Authorization Server Support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/330200b4d8c6558a8ea2964d521b5d2f659a48c0"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:22:07.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-11-20 20:22:09.000000000","tz":-180},"subject":"Add schema version and support to \"domain\" attribute in mapping rules","message":"Add schema version and support to \"domain\" attribute in mapping rules\n\nAs discussed in the Keystone meeting [1], I am proposing this spec, so\nwe can review and discuss the implementation proposed in [2].\n\n[1] http://eavesdrop.openstack.org/meetings/keystone/2020/keystone.2020-08-25-16.59.log.txt\n[2] https://review.opendev.org/#/c/739966/\n\nChange-Id: I84426119a60c595eaf462662c6a24a990914c516\nImplements: blueprint versioning-for-attribute-mapping-schema\nRFE: https://bugs.launchpad.net/keystone/+bug/1887515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/62bcfa6f3fa28ed4e058c372db7646ca0110b409"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/62bcfa6f3fa28ed4e058c372db7646ca0110b409"}]},"branch":"refs/heads/master"},"2170b9176d6d8edaecb2bfee998f9a2d082b180c":{"kind":"REWORK","_number":17,"created":"2023-11-20 20:41:01.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/42/748042/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/42/748042/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/42/748042/17"}}},"commit":{"parents":[{"commit":"330200b4d8c6558a8ea2964d521b5d2f659a48c0","subject":"Merge \"External OAuth2.0 Authorization Server Support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/330200b4d8c6558a8ea2964d521b5d2f659a48c0"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-25 22:22:07.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2023-11-20 20:40:45.000000000","tz":-180},"subject":"Add schema version and support to \"domain\" attribute in mapping rules","message":"Add schema version and support to \"domain\" attribute in mapping rules\n\nAs discussed in the Keystone meeting [1], I am proposing this spec, so\nwe can review and discuss the implementation proposed in [2].\n\n[1] http://eavesdrop.openstack.org/meetings/keystone/2020/keystone.2020-08-25-16.59.log.txt\n[2] https://review.opendev.org/#/c/739966/\n\nChange-Id: I84426119a60c595eaf462662c6a24a990914c516\nImplements: blueprint versioning-for-attribute-mapping-schema\nRFE: https://bugs.launchpad.net/keystone/+bug/1887515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/2170b9176d6d8edaecb2bfee998f9a2d082b180c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/2170b9176d6d8edaecb2bfee998f9a2d082b180c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
