)]}'
{"id":"openstack%2Fkeystone-specs~787640","triplet_id":"openstack%2Fkeystone-specs~master~Ice3ffd025a706a5e3c1c450bbe9813ac8f558a9e","project":"openstack/keystone-specs","branch":"master","hashtags":[],"change_id":"Ice3ffd025a706a5e3c1c450bbe9813ac8f558a9e","subject":"[spec] X-Project-Id Pass-through","status":"MERGED","created":"2021-04-22 21:23:47.000000000","updated":"2021-08-03 15:57:18.000000000","submitted":"2021-08-03 15:55:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":41,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"787640","meta_rev_id":"5637b36ff8cea7679e2b45d926f72b9ce64cf55d","_number":787640,"virtual_id_number":787640,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-03 15:55:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2021-06-08 14:40:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2021-08-02 21:16:25.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":1,"date":"2021-06-25 06:49:05.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-24 15:34:55.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2021-08-03 15:44:35.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":31661,"name":"Dave Wilde","username":"dwilde","inactive":true}],"reviewers":{"CC":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31661,"name":"Dave Wilde","username":"dwilde","inactive":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-22 21:24:05.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-04-22 22:01:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-23 14:37:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-04-23 19:19:07.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-04-23 19:19:07.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-04-26 20:45:26.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2021-05-10 12:08:31.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2021-06-04 16:46:44.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":31661,"name":"Dave Wilde","username":"dwilde","inactive":true},"state":"REVIEWER"},{"updated":"2021-06-04 17:42:56.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2021-06-14 15:33:33.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2021-06-14 15:33:33.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"}],"messages":[{"id":"f59c03e0cb1e48b6d4200134250a40cd7bb230c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-04-22 21:23:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b9a4dc4e29b36474e53d63a41ef61fab9be3772","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-04-22 21:24:05.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"c68d33db3141ba63622c9a7d65efc13d1e7a9f6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 22:01:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3625d49b2b0e4a0c89cc905d114aa031 : FAILURE in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ce18eb47ebc43c2beabe652b8bbcc97 : FAILURE in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"f35b78c7ef2309e27e446524be62930e0e4a3f8f","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-04-23 19:29:19.000000000","message":"Patch Set 1:\n\n(9 comments)\n\nThanks for taking the time to write this up, Doug.","accounts_in_message":[],"_revision_number":1},{"id":"f3d257c0003bddaa09779e1b1cb99577ec206966","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-04-23 21:41:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04df3d17063722b96648c76a8b39409bcb5a608c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-04-24 11:08:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)\n\nactully i quite like this.\nill think about it over the weekend and take another look on monday but\nthis would adress many of the issue i have with the current approch.\n\nim still not sure this solve all cases for example if operators create addtional\nroles and use those in policy files  but dont grant them to the system admin tokens\nthen my system user can operate on it.\n\ni think though this addresses enough of the usability issue that we could see how it went\nand maybe cross that bridge when we came to it.\n\nthe -1 is rather minor in that i would prefer if you did not call the retention of the headder caching since its not really caching, you storing it for retrival later which is similar but not the same.\ni would have left it as  just a nit if it was not for the fact there are other comments and you have a -w on the patch yourself so i assume your intended to respin it anyway based on feedback.","accounts_in_message":[],"_revision_number":1},{"id":"8051637b6acdcd54683a5d25f6e174bdc23117e3","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-04-26 12:59:50.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4ec3ba1c8e9a084b51f6d1cb0a4ba3a01cb5980c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-26 20:45:26.000000000","message":"Patch Set 1:\n\nI like that idea, thx for writing it :)","accounts_in_message":[],"_revision_number":1},{"id":"07bdf1e2af3cdd6e1d9e1ee510778a7ee4d801ad","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-04-29 00:01:22.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nthis is a rather soft -1 by the way i do quite like this idea","accounts_in_message":[],"_revision_number":1},{"id":"a0fe317d7389c55d54a981b63ff8413d14325290","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-05-10 12:16:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"714d1e59a2eab4e3ea68257fd9d8e85d0932effb","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-05-12 13:22:47.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6114f330133715bf28bb3d3afd672c4eaffde4dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-06-04 16:45:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bf87b8dcab1fed38830f1c04d0e173aca4d0e037","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-06-04 16:46:03.000000000","message":"Patch Set 2:\n\n(8 comments)\n\nRevised spec to address comments.","accounts_in_message":[],"_revision_number":2},{"id":"d580f58a742dee26afdb44a7c7e01dcd4c57665d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 16:54:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/676b481c030d4888a97b835effab31e3 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e96b4e7f66354abb82945367f03719f3 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":2},{"id":"d156746b0bfccb689db692b039b2810e02064523","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-06-04 17:42:56.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"46671ce6c8712f2ee7b553ea07852fb1fbfdb34a","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-06-04 17:43:15.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"a876754fc613d4a69e95a0e98624225731aed8d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-06-04 18:02:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b60a1de100231097180bbcf93de397e969e86e0a","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-06-04 18:02:36.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5f7f6db5ae234a5243fc0eba4d2e14be0d9fec85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 18:12:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c727a2f3803f45309c31dbda829a613f : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ad2aa6aac8e42b0bf6e9c6c50ba482e : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":3},{"id":"d84c5b9b65582e6987e680a47f14f22fa4db97bf","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-06-08 14:40:43.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)\n\n+1 from an end user experience point of view and the fact this proposal limits/elimiates the impact for service project that expect a project id.","accounts_in_message":[],"_revision_number":3},{"id":"79999899cad7e1e849d2379b85a74556e96840a3","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-06-08 18:05:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5266534dc50bf2bf661367ed2b961e462a4685b3","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2021-06-08 19:18:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f8e0b334ebd119508a78c90197e59bcaa85b6744","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2021-06-24 15:34:55.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"aa776f12598e3b926497731f83cb91bd5da82c46","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-06-25 06:49:05.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1f2c4176b2e7da9b40e6dbc1ee3b9f759fa87e05","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-02 21:16:25.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c65a9e47252979fe81c3aeb6ccbfb1747b3ae409","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-03 15:44:35.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"79ee80b115dda4c0cbb3344561a62f004504050b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 15:44:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"41d2417bf39d416ae54445b690aa1ac67dd253b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 15:55:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1db42e80e27843b39132e43be1601ea4 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1551e52bb1074061a8e830fff86fbc49 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":3},{"id":"4a06556242046b09182638d461b286ccda0c12ed","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 15:55:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5637b36ff8cea7679e2b45d926f72b9ce64cf55d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 15:57:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/b4d97b30fa82445bb215ba2129783481 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"18760db9fcdf3eb2cccb2b11f8503acb0d003d2b","revisions":{"7b73cdd9e7deb5511160905cc14496de44ff73d1":{"kind":"REWORK","_number":1,"created":"2021-04-22 21:23:47.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/40/787640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/40/787640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/40/787640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/40/787640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/40/787640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/40/787640/1"}}},"commit":{"parents":[{"commit":"70f9caa74060171df8eb63b2688f88b7fce17184","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/70f9caa74060171df8eb63b2688f88b7fce17184"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-04-22 21:23:10.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-04-22 21:23:46.000000000","tz":-300},"subject":"WIP [spec] X-Project-Id Pass-through","message":"WIP [spec] X-Project-Id Pass-through\n\nChange-Id: Ice3ffd025a706a5e3c1c450bbe9813ac8f558a9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/7b73cdd9e7deb5511160905cc14496de44ff73d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/7b73cdd9e7deb5511160905cc14496de44ff73d1"}]},"branch":"refs/heads/master"},"74bc8de42ccb880c3e1510b2b8803de03d3983f2":{"kind":"REWORK","_number":2,"created":"2021-06-04 16:45:02.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/40/787640/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/40/787640/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/40/787640/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/40/787640/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/40/787640/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/40/787640/2"}}},"commit":{"parents":[{"commit":"70f9caa74060171df8eb63b2688f88b7fce17184","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/70f9caa74060171df8eb63b2688f88b7fce17184"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-04-22 21:23:10.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-06-04 16:44:53.000000000","tz":-300},"subject":"[spec] X-Project-Id Pass-through","message":"[spec] X-Project-Id Pass-through\n\nChange-Id: Ice3ffd025a706a5e3c1c450bbe9813ac8f558a9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/74bc8de42ccb880c3e1510b2b8803de03d3983f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/74bc8de42ccb880c3e1510b2b8803de03d3983f2"}]},"branch":"refs/heads/master"},"18760db9fcdf3eb2cccb2b11f8503acb0d003d2b":{"kind":"REWORK","_number":3,"created":"2021-06-04 18:02:31.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/40/787640/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/40/787640/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/40/787640/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/40/787640/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/40/787640/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/40/787640/3"}}},"commit":{"parents":[{"commit":"70f9caa74060171df8eb63b2688f88b7fce17184","subject":"Switch to newer openstackdocstheme version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/70f9caa74060171df8eb63b2688f88b7fce17184"}]}],"author":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-04-22 21:23:10.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2021-06-04 18:02:26.000000000","tz":-300},"subject":"[spec] X-Project-Id Pass-through","message":"[spec] X-Project-Id Pass-through\n\nChange-Id: Ice3ffd025a706a5e3c1c450bbe9813ac8f558a9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18760db9fcdf3eb2cccb2b11f8503acb0d003d2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18760db9fcdf3eb2cccb2b11f8503acb0d003d2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
