)]}'
{"id":"openstack%2Fkeystone-specs~813152","triplet_id":"openstack%2Fkeystone-specs~master~I4954c1e8f22199deb13031441c46a3565383412d","project":"openstack/keystone-specs","branch":"master","topic":"bp/oauth2-client-credentials-ext","attention_set":{},"removed_from_attention_set":{"33920":{"account":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"last_update":"2022-02-15 19:26:25.000000000","reason":"Change was submitted"},"16465":{"account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"last_update":"2022-02-08 15:05:18.000000000","reason":"Kristi Nikolla replied on the change"},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2022-02-09 18:21:18.000000000","reason":"Yasufumi Ogawa replied on the change"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2021-10-12 15:12:34.000000000","reason":"Douglas Mendizábal replied on the change"},"5046":{"account":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"last_update":"2022-02-15 19:26:25.000000000","reason":"Change was submitted"},"2218":{"account":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"last_update":"2022-02-15 19:26:25.000000000","reason":"Change was submitted"},"32604":{"account":{"_account_id":32604,"name":"Keiichiro Yamakawa","email":"yamakawa.keiich@fujitsu.com"},"last_update":"2022-02-15 19:26:25.000000000","reason":"Change was submitted"},"21420":{"account":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"last_update":"2022-02-15 19:19:41.000000000","reason":"Gage Hugo replied on the change"},"33455":{"account":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"last_update":"2022-02-15 19:26:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4954c1e8f22199deb13031441c46a3565383412d","subject":"OAuth2.0 Client Credentials Grant Flow Support","status":"MERGED","created":"2021-10-08 09:09:52.000000000","updated":"2022-02-24 05:41:47.000000000","submitted":"2022-02-15 19:26:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":82,"unresolved_comment_count":8,"has_review_started":true,"submission_id":"813152-bp/oauth2-client-credentials-ext","meta_rev_id":"34e441bfa130d851f4345f32c4de1f28cbf95987","_number":813152,"virtual_id_number":813152,"owner":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-15 19:26:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-02-08 15:05:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2022-02-09 18:21:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-15 19:19:41.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-15 19:19:41.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"_account_id":32604,"name":"Keiichiro Yamakawa","email":"yamakawa.keiich@fujitsu.com"}],"reviewers":{"REVIEWER":[{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":32604,"name":"Keiichiro Yamakawa","email":"yamakawa.keiich@fujitsu.com"},{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-08 09:20:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-12 15:09:01.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2021-10-12 15:12:34.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":2218,"name":"Adam Young","email":"adam@younglogic.com","username":"ayoung"},"state":"REVIEWER"},{"updated":"2021-10-19 03:46:51.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":32604,"name":"Keiichiro Yamakawa","email":"yamakawa.keiich@fujitsu.com"},"state":"REVIEWER"},{"updated":"2021-10-19 03:46:51.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-10-19 03:46:51.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2021-10-19 03:46:51.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2021-12-16 01:34:01.000000000","updated_by":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"reviewer":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"state":"REVIEWER"},{"updated":"2022-01-19 01:54:27.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"}],"messages":[{"id":"71e74c9d288fa481d2f9b3372cd92f28c95b1756","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2021-10-08 09:09:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c04f1031dcc2efd074f45149aa4f94c628039cb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-08 09:20:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b798f1fde054d13afc0cfe8569f976f : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3007322455c483c975c35075967dc86 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de705e3dbe594510aa66562b1d227bc6 : SUCCESS in 4m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc2afb3bdb6e415dab5a8ef79e4e3422 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d527945f1f6b4929a7f6d7b91cd53c77 : SUCCESS in 5m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"413b5e946eb58872c8058c7cd40b777fd6d19e21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2021-10-11 02:43:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e3d83f3066e18a085d58f21cc50643f407f0e012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 02:54:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e20d4b4e680c44348cd6e1f7495356c4 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a36f12252ff24f1cb71bf2a95e99cebf : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9441ee8c56154b91b663754adb775fff : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d3db4cdb78444034b07cb083b7977819 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9f2929435e34ac0a15f2d28d9188d75 : SUCCESS in 4m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3eb1cb8e63446df77fb28c1c96665a0a6b5d03a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2021-10-12 09:54:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c5ce502e7a27416670ca103195f39c783f01f700","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 10:11:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eea98cdcfe5143e7ac1e05ca232e0302 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c788ef6398094b51a608f3418c835d6d : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3129bc27f6641289a6b28aea6ddfd31 : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/617226312d4947b08384a6974a7728d1 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5101200456d24eccb7ee1067b4dc9248 : SUCCESS in 3m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1a0c22fd215da859697f9be6eda585654fbfc0c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2021-10-12 14:57:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"92029b4f40a0e15d447b73a0dbd3cfa8eed0215e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 15:05:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ebade28b03848b6be79c0b41b74f046 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2587b23fcb5a4fcb860d145bea80cefc : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa84c1690775467fae1964319aa1b93e : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc7124a39ce44cedbd5f12190c65d379 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11b2d60d6cbe45b78689c8d7f7685828 : SUCCESS in 4m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8cba751ef2a08475ab350bf496502b38b37ba19d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2021-10-19 03:44:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"263f0c0ec5e79230cf4caf52fd1f2f2c9ca95fb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 03:54:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85ab8c5409be4660b36506863530fd41 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b748e8c44434c6b829cc225a067b0e8 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f721ec126db42399422f64ac468bfb7 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b1880581bec483881f898c9d39ed258 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1810d5e9ab0d4a17b12defcbd366964c : SUCCESS in 3m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d37c32fbe59695b9d479787393b47d40606efe6e","author":{"_account_id":32604,"name":"Keiichiro Yamakawa","email":"yamakawa.keiich@fujitsu.com"},"date":"2021-10-20 09:23:54.000000000","message":"Patch Set 5:\n\n(17 comments)","accounts_in_message":[],"_revision_number":5},{"id":"9d24c32849a0cb74f64d3c505fe2717869de8796","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2021-10-21 03:04:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"80efdcf5f19addea891bc44fa836857b9c61bc02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-21 03:18:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f81467a373f24666accdf8b0822de415 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/170b81ea82cd4ce9a06d0a8861f2daa6 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3aced16d87e47cebf5f670865756350 : SUCCESS in 6m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4260fe287dce477b9a3f72db31ed5234 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bbb33569f354405af5888bff5f1cd37 : SUCCESS in 6m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"37b2fb72310c50ccbecbb58ad9dd3d76a4fdc2b7","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2021-10-21 03:22:35.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8fcaf0df1d12f8d38e1f1d4d6bd63f5de1030022","author":{"_account_id":32604,"name":"Keiichiro Yamakawa","email":"yamakawa.keiich@fujitsu.com"},"date":"2021-10-22 01:56:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"824af2db50acde4364b15a815829cd3b8b9ec835","author":{"_account_id":32604,"name":"Keiichiro Yamakawa","email":"yamakawa.keiich@fujitsu.com"},"date":"2021-10-22 02:23:45.000000000","message":"Patch Set 6:\n\n(18 comments)","accounts_in_message":[],"_revision_number":6},{"id":"d5e98d5eb859e735dceb8573b2d9394b822e4bb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2021-10-25 03:01:18.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"c1087fa09d081743991100e3c61bd03c62f399bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-25 03:10:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fff78aeeb90240d09d58286bb6d24360 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17f1462be809466488f898edad8fe42d : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af93343cb4e74d62bd2644d8dd4d919f : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/374852b39dbb4cf9bf43cf12799bdcb8 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/247cd8b239b648ac82fd5095f3cc61eb : SUCCESS in 6m 05s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3f270e33613b5b46e10df6e3f71c729bf70f57e1","author":{"_account_id":32604,"name":"Keiichiro Yamakawa","email":"yamakawa.keiich@fujitsu.com"},"date":"2021-10-28 01:44:29.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3b68fc68c7d00c73699c17a4c143f491105b6501","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2021-11-02 14:03:54.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"66d5216b1ae83362f023dbe82d2c7ca6392f2441","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2021-11-10 08:33:20.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7deb6d4158728298a026aebfd1d717a85543cb58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 08:43:00.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d7f7105455a4574b96f938f103c0d8b : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffd79270a4204d25a19d7b2e5d7826a5 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4955a93d1824482ca73792d5b903e04f : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1710edfb998c42a687068d291e93aa8b : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bfca083a9be4e5fb38ac0b0ba58457a : SUCCESS in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d265f1b5d3cbc49ab514e826de8f5374b1cfb38d","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2021-11-10 09:44:19.000000000","message":"Patch Set 8:\n\n(21 comments)","accounts_in_message":[],"_revision_number":8},{"id":"79564f377b21d27101901a8df661755f0ad72afe","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2021-11-18 15:59:42.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"6607b647fa5f67131f5f28b87e67d00cc65525c5","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2021-11-30 14:49:21.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"b5f76e9aeb7a5eadc8d6892b300f8cd0ebb09637","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2021-12-08 02:52:35.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4017ffc33e13243d68d32d095c1e3285c18dce30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-08 03:08:49.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be7d70498f1c49f69f3979bbbd15e169 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4a3adc88bff418fa2bdb98d59504311 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/432faf485db745809110cc5165b32338 : SUCCESS in 6m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2fa199eb4c041bab9b0d0b71d1d55f3 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28a1d6e8680140d7a85ed7feea4e7b5c : SUCCESS in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e9432b298e0ad66e94e4bf52e1b5f6a0c7db4b76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-01-06 06:43:46.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"68e80168a06ddbcef3ba5e453b81fc9ef8f4a149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 06:51:00.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce13730e171d49e3854f601b6aed3f3f : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5a7ccf8807140e382774b5c1b76209b : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/885b73c493e04bff9005de7f60ac8b83 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2204d180ee304344826ee5b1422cd71a : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44821ac7f0954fd3872905a30d3390d9 : SUCCESS in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"cc85d17d85d4ba834befe30013d7aaf59b6fed32","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-01-06 06:55:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9f409a72d4d8f5da079331819c4adacaf97b096d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-01-14 10:28:19.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"3a9ade435605ca4f392578d0eceabdaf3916cca4","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-01-14 10:34:02.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"def677284dc7edd76ed305d41b7af6f40677155b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 10:38:42.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/176a56811a104d49878e8469df4195ca : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8f17b3c069e4238b8f2e01cff1208d9 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d6cbc675d3d4bce93c6a5549ac57236 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1cb2eb8f75784431a45dd8b32ca9c498 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/901c52197b42471b8b922cf3b5305e6a : SUCCESS in 3m 14s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5055cd979b4e98d472156540b9d2d625fcc45e82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-01-24 10:21:50.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"37c8e0f8a8227ef410082f114251cd57f968baa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-24 10:30:55.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0388d47079740acbdcab46d2ead0da7 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d555d951679b41349c7679e5e4a96af0 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64f71174109946fa9b1bcc3c2584959d : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2938d78c31e4d6badf7bdb746fddf53 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ca06813ad284dff82a743864ddb69f5 : SUCCESS in 4m 10s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"24e949f60a5768569c818123612ce6ff38a85131","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-01-25 15:01:39.000000000","message":"Patch Set 12: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"c1712816d91019f4730d61addb8111d5b77e8881","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-01-27 10:05:25.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"ba5dc443384d0c2483551051e2eb2ce1df42b477","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 10:19:10.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ce83f04c16d47469a0042a67ef9de40 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a20f529a478e49b3aba68a790f7dfd11 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/087d2b8db0234345837fdb647ae40d5a : SUCCESS in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9eea9196f57749818c3ba32700e7b379 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5146e27b7110449a9668d8caf81a3462 : SUCCESS in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"c297ca2176f0cef3ec82e19765b323b6cbd8e581","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-01-28 05:21:02.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"9f261950d7c8a5b00b203bc4bf167828ae93e3dd","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-01-28 05:25:40.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"e9873020ce3e70e210636be9050f56f19897a3a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 05:30:11.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/532991c2441d4512abf2fc83ca1257e5 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58f3779b361d4f60ab9b963f056c79ae : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f889ea00fcb54695aba49ae93fac6277 : SUCCESS in 3m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/81c83001f1044d41b73194ca139b3f94 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/095d5d60bbc943a6a3139d55cb6bd177 : SUCCESS in 3m 10s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"56e1b21cf9a654de4019b7cb3f80f3284ee69bf3","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-02-08 15:05:18.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"8efd7af1061b007fa27a095fd808db8afad674d7","author":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"date":"2022-02-09 18:21:18.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"dbae8e8188fdbf1db59f963e8804971923f6eca7","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-02-15 19:19:41.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"93bb7a5e9bf659882e811a2e5e6f4d2d4bb9466c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 19:20:14.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"d492903ac947f4edc532bca670417f44a07ef40f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 19:26:25.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/031c4c2cbba84d579e61260781568333 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9fe8b522c044bc5a96fe12c3fa074ef : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4640742256224e0b93c846673fac5a09 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c66e44bd455244b7ac71b8acec3cc952 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":14},{"id":"e02e84fdfebd1da86596abd3a601370520eab6ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 19:26:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"da57843943d44f78b31ef1d3ec3e5f4d508f4f98","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 19:28:07.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/a31da4e004054ed380163f49ad0b877e : SUCCESS in 54s","accounts_in_message":[],"_revision_number":14},{"id":"34e441bfa130d851f4345f32c4de1f28cbf95987","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-02-24 05:41:47.000000000","message":"Patch Set 14:\n\n(3 comments)","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"8145886d241a48e9dd418c4b56a3a200a0ffe736","revisions":{"b364b652d3941b33874f2172b4003fd83b0bd6cd":{"kind":"REWORK","_number":1,"created":"2021-10-08 09:09:52.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/52/813152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/52/813152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/1"}}},"commit":{"parents":[{"commit":"18760db9fcdf3eb2cccb2b11f8503acb0d003d2b","subject":"[spec] X-Project-Id Pass-through","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18760db9fcdf3eb2cccb2b11f8503acb0d003d2b"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-08 08:59:29.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-08 09:09:21.000000000","tz":540},"subject":"OAuth2.0 Client Credentials Grant Flow Support","message":"OAuth2.0 Client Credentials Grant Flow Support\n\nThis spec proposes to allow users to optionally use an OAuth2.0 Client\nCredentials Grant flow to authorize an API client. In order to realize\nthis, we implement an OAuth2.0 authorization server as an extension of\nkeystone.\n\nImplements: blueprint oauth2-client-credentials-ext\nChange-Id: I4954c1e8f22199deb13031441c46a3565383412d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b364b652d3941b33874f2172b4003fd83b0bd6cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b364b652d3941b33874f2172b4003fd83b0bd6cd"}]},"branch":"refs/heads/master"},"b59ddf75ae1d346417865da02366cd963ca03d99":{"kind":"REWORK","_number":2,"created":"2021-10-11 02:43:30.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/52/813152/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/52/813152/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/2"}}},"commit":{"parents":[{"commit":"18760db9fcdf3eb2cccb2b11f8503acb0d003d2b","subject":"[spec] X-Project-Id Pass-through","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18760db9fcdf3eb2cccb2b11f8503acb0d003d2b"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-08 08:59:29.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-11 02:43:18.000000000","tz":540},"subject":"OAuth2.0 Client Credentials Grant Flow Support","message":"OAuth2.0 Client Credentials Grant Flow Support\n\nThis spec proposes to allow users to optionally use an OAuth2.0 Client\nCredentials Grant flow to authorize an API client. In order to realize\nthis, we implement an OAuth2.0 authorization server as an extension of\nkeystone.\n\nImplements: blueprint oauth2-client-credentials-ext\nChange-Id: I4954c1e8f22199deb13031441c46a3565383412d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b59ddf75ae1d346417865da02366cd963ca03d99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b59ddf75ae1d346417865da02366cd963ca03d99"}]},"branch":"refs/heads/master"},"b1865805b05e6dc9459d0fb39d22688710b3cadc":{"kind":"REWORK","_number":3,"created":"2021-10-12 09:54:48.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/52/813152/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/52/813152/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/3"}}},"commit":{"parents":[{"commit":"18760db9fcdf3eb2cccb2b11f8503acb0d003d2b","subject":"[spec] X-Project-Id Pass-through","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18760db9fcdf3eb2cccb2b11f8503acb0d003d2b"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-08 08:59:29.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-12 09:54:02.000000000","tz":540},"subject":"OAuth2.0 Client Credentials Grant Flow Support","message":"OAuth2.0 Client Credentials Grant Flow Support\n\nThis spec proposes to allow users to optionally use an OAuth2.0 Client\nCredentials Grant flow to authorize an API client. In order to realize\nthis, we implement an OAuth2.0 authorization server as an extension of\nkeystone.\n\nImplements: blueprint oauth2-client-credentials-ext\nChange-Id: I4954c1e8f22199deb13031441c46a3565383412d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b1865805b05e6dc9459d0fb39d22688710b3cadc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b1865805b05e6dc9459d0fb39d22688710b3cadc"}]},"branch":"refs/heads/master"},"c4cbee5c69204ca23253b6507a438d28f0b361a2":{"kind":"REWORK","_number":4,"created":"2021-10-12 14:57:40.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/52/813152/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/52/813152/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/4"}}},"commit":{"parents":[{"commit":"18760db9fcdf3eb2cccb2b11f8503acb0d003d2b","subject":"[spec] X-Project-Id Pass-through","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18760db9fcdf3eb2cccb2b11f8503acb0d003d2b"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-08 08:59:29.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-12 14:56:32.000000000","tz":540},"subject":"OAuth2.0 Client Credentials Grant Flow Support","message":"OAuth2.0 Client Credentials Grant Flow Support\n\nThis spec proposes to allow users to optionally use an OAuth2.0 Client\nCredentials Grant flow to authorize an API client. In order to realize\nthis, we implement an OAuth2.0 authorization server as an extension of\nkeystone.\n\nImplements: blueprint oauth2-client-credentials-ext\nChange-Id: I4954c1e8f22199deb13031441c46a3565383412d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/c4cbee5c69204ca23253b6507a438d28f0b361a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/c4cbee5c69204ca23253b6507a438d28f0b361a2"}]},"branch":"refs/heads/master"},"f285b78e4dbd6a608f72075c7a85e17c0c1471b7":{"kind":"REWORK","_number":5,"created":"2021-10-19 03:44:50.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/52/813152/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/52/813152/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/5"}}},"commit":{"parents":[{"commit":"18760db9fcdf3eb2cccb2b11f8503acb0d003d2b","subject":"[spec] X-Project-Id Pass-through","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18760db9fcdf3eb2cccb2b11f8503acb0d003d2b"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-08 08:59:29.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-19 03:44:11.000000000","tz":540},"subject":"OAuth2.0 Client Credentials Grant Flow Support","message":"OAuth2.0 Client Credentials Grant Flow Support\n\nThis spec proposes to allow users to optionally use an OAuth2.0 Client\nCredentials Grant flow to authorize an API client. In order to realize\nthis, we implement an OAuth2.0 authorization server as an extension of\nkeystone.\n\nImplements: blueprint oauth2-client-credentials-ext\nChange-Id: I4954c1e8f22199deb13031441c46a3565383412d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f285b78e4dbd6a608f72075c7a85e17c0c1471b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f285b78e4dbd6a608f72075c7a85e17c0c1471b7"}]},"branch":"refs/heads/master"},"a716ee64b236aa4504aa9ed17a65bd9f83ef9344":{"kind":"REWORK","_number":6,"created":"2021-10-21 03:04:35.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/52/813152/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/52/813152/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/6"}}},"commit":{"parents":[{"commit":"18760db9fcdf3eb2cccb2b11f8503acb0d003d2b","subject":"[spec] X-Project-Id Pass-through","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18760db9fcdf3eb2cccb2b11f8503acb0d003d2b"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-08 08:59:29.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-21 03:04:15.000000000","tz":540},"subject":"OAuth2.0 Client Credentials Grant Flow Support","message":"OAuth2.0 Client Credentials Grant Flow Support\n\nThis spec proposes to allow users to optionally use an OAuth2.0 Client\nCredentials Grant flow to authorize an API client. In order to realize\nthis, we implement an OAuth2.0 authorization server as an extension of\nkeystone.\n\nImplements: blueprint oauth2-client-credentials-ext\nChange-Id: I4954c1e8f22199deb13031441c46a3565383412d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/a716ee64b236aa4504aa9ed17a65bd9f83ef9344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/a716ee64b236aa4504aa9ed17a65bd9f83ef9344"}]},"branch":"refs/heads/master"},"0ba88c616250931861545c600dbcb72cf7936143":{"kind":"REWORK","_number":7,"created":"2021-10-25 03:01:18.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/52/813152/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/52/813152/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/7"}}},"commit":{"parents":[{"commit":"18760db9fcdf3eb2cccb2b11f8503acb0d003d2b","subject":"[spec] X-Project-Id Pass-through","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18760db9fcdf3eb2cccb2b11f8503acb0d003d2b"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-08 08:59:29.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-25 03:01:05.000000000","tz":540},"subject":"OAuth2.0 Client Credentials Grant Flow Support","message":"OAuth2.0 Client Credentials Grant Flow Support\n\nThis spec proposes to allow users to optionally use an OAuth2.0 Client\nCredentials Grant flow to authorize an API client. In order to realize\nthis, we implement an OAuth2.0 authorization server as an extension of\nkeystone.\n\nImplements: blueprint oauth2-client-credentials-ext\nChange-Id: I4954c1e8f22199deb13031441c46a3565383412d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/0ba88c616250931861545c600dbcb72cf7936143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/0ba88c616250931861545c600dbcb72cf7936143"}]},"branch":"refs/heads/master"},"c6bd4f232c0ca79c2313260aeb23a1d9f23ac83b":{"kind":"REWORK","_number":8,"created":"2021-11-10 08:33:20.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/52/813152/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/52/813152/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/8"}}},"commit":{"parents":[{"commit":"18760db9fcdf3eb2cccb2b11f8503acb0d003d2b","subject":"[spec] X-Project-Id Pass-through","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18760db9fcdf3eb2cccb2b11f8503acb0d003d2b"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-08 08:59:29.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-11-10 08:32:53.000000000","tz":540},"subject":"OAuth2.0 Client Credentials Grant Flow Support","message":"OAuth2.0 Client Credentials Grant Flow Support\n\nThis spec proposes to allow users to optionally use an OAuth2.0 Client\nCredentials Grant flow to authorize an API client. In order to realize\nthis, we implement an OAuth2.0 authorization server as an extension of\nkeystone.\n\nImplements: blueprint oauth2-client-credentials-ext\nChange-Id: I4954c1e8f22199deb13031441c46a3565383412d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/c6bd4f232c0ca79c2313260aeb23a1d9f23ac83b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/c6bd4f232c0ca79c2313260aeb23a1d9f23ac83b"}]},"branch":"refs/heads/master"},"b16b904c6d674f760c288461e379ca80e5fc5fe1":{"kind":"REWORK","_number":9,"created":"2021-12-08 02:52:35.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/52/813152/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/52/813152/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/9"}}},"commit":{"parents":[{"commit":"18760db9fcdf3eb2cccb2b11f8503acb0d003d2b","subject":"[spec] X-Project-Id Pass-through","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18760db9fcdf3eb2cccb2b11f8503acb0d003d2b"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-08 08:59:29.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-12-08 02:52:18.000000000","tz":540},"subject":"OAuth2.0 Client Credentials Grant Flow Support","message":"OAuth2.0 Client Credentials Grant Flow Support\n\nThis spec proposes to allow users to optionally use an OAuth2.0 Client\nCredentials Grant flow to authorize an API client. In order to realize\nthis, we implement an OAuth2.0 authorization server as an extension of\nkeystone.\n\nImplements: blueprint oauth2-client-credentials-ext\nChange-Id: I4954c1e8f22199deb13031441c46a3565383412d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b16b904c6d674f760c288461e379ca80e5fc5fe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b16b904c6d674f760c288461e379ca80e5fc5fe1"}]},"branch":"refs/heads/master"},"1b2884c8d9b595445fad8cd91c284bd1e0ffc0fa":{"kind":"REWORK","_number":10,"created":"2022-01-06 06:43:46.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/52/813152/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/52/813152/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/10"}}},"commit":{"parents":[{"commit":"18760db9fcdf3eb2cccb2b11f8503acb0d003d2b","subject":"[spec] X-Project-Id Pass-through","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18760db9fcdf3eb2cccb2b11f8503acb0d003d2b"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-08 08:59:29.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-01-06 06:43:22.000000000","tz":540},"subject":"OAuth2.0 Client Credentials Grant Flow Support","message":"OAuth2.0 Client Credentials Grant Flow Support\n\nThis spec proposes to allow users to optionally use an OAuth2.0 Client\nCredentials Grant flow to authorize an API client. In order to realize\nthis, we implement an OAuth2.0 authorization server as an extension of\nkeystone.\n\nImplements: blueprint oauth2-client-credentials-ext\nChange-Id: I4954c1e8f22199deb13031441c46a3565383412d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/1b2884c8d9b595445fad8cd91c284bd1e0ffc0fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/1b2884c8d9b595445fad8cd91c284bd1e0ffc0fa"}]},"branch":"refs/heads/master"},"b503863d93460a3522c6c3318af192f88429286d":{"kind":"REWORK","_number":11,"created":"2022-01-14 10:28:19.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/52/813152/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/52/813152/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/11"}}},"commit":{"parents":[{"commit":"18760db9fcdf3eb2cccb2b11f8503acb0d003d2b","subject":"[spec] X-Project-Id Pass-through","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18760db9fcdf3eb2cccb2b11f8503acb0d003d2b"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-08 08:59:29.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-01-14 10:28:09.000000000","tz":540},"subject":"OAuth2.0 Client Credentials Grant Flow Support","message":"OAuth2.0 Client Credentials Grant Flow Support\n\nThis spec proposes to allow users to optionally use an OAuth2.0 Client\nCredentials Grant flow to authorize an API client. In order to realize\nthis, we implement an OAuth2.0 authorization server as an extension of\nkeystone.\n\nImplements: blueprint oauth2-client-credentials-ext\nChange-Id: I4954c1e8f22199deb13031441c46a3565383412d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b503863d93460a3522c6c3318af192f88429286d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b503863d93460a3522c6c3318af192f88429286d"}]},"branch":"refs/heads/master"},"da9aa04912f0042f2f306f2ebb103612394fe280":{"kind":"REWORK","_number":12,"created":"2022-01-24 10:21:50.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/52/813152/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/52/813152/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/12"}}},"commit":{"parents":[{"commit":"18760db9fcdf3eb2cccb2b11f8503acb0d003d2b","subject":"[spec] X-Project-Id Pass-through","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18760db9fcdf3eb2cccb2b11f8503acb0d003d2b"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-08 08:59:29.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-01-24 10:21:37.000000000","tz":540},"subject":"OAuth2.0 Client Credentials Grant Flow Support","message":"OAuth2.0 Client Credentials Grant Flow Support\n\nThis spec proposes to allow users to optionally use an OAuth2.0 Client\nCredentials Grant flow to authorize an API client. In order to realize\nthis, we implement an OAuth2.0 authorization server as an extension of\nkeystone.\n\nImplements: blueprint oauth2-client-credentials-ext\nChange-Id: I4954c1e8f22199deb13031441c46a3565383412d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/da9aa04912f0042f2f306f2ebb103612394fe280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/da9aa04912f0042f2f306f2ebb103612394fe280"}]},"branch":"refs/heads/master"},"424413eb2fd4646f2f458b33ee89fb02b6f5b912":{"kind":"REWORK","_number":13,"created":"2022-01-27 10:05:25.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/52/813152/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/52/813152/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/13"}}},"commit":{"parents":[{"commit":"18760db9fcdf3eb2cccb2b11f8503acb0d003d2b","subject":"[spec] X-Project-Id Pass-through","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18760db9fcdf3eb2cccb2b11f8503acb0d003d2b"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-08 08:59:29.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-01-27 10:05:13.000000000","tz":540},"subject":"OAuth2.0 Client Credentials Grant Flow Support","message":"OAuth2.0 Client Credentials Grant Flow Support\n\nThis spec proposes to allow users to optionally use an OAuth2.0 Client\nCredentials Grant flow to authorize an API client. In order to realize\nthis, we implement an OAuth2.0 authorization server as an extension of\nkeystone.\n\nImplements: blueprint oauth2-client-credentials-ext\nChange-Id: I4954c1e8f22199deb13031441c46a3565383412d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/424413eb2fd4646f2f458b33ee89fb02b6f5b912"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/424413eb2fd4646f2f458b33ee89fb02b6f5b912"}]},"branch":"refs/heads/master"},"8145886d241a48e9dd418c4b56a3a200a0ffe736":{"kind":"REWORK","_number":14,"created":"2022-01-28 05:21:02.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/52/813152/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/52/813152/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/52/813152/14"}}},"commit":{"parents":[{"commit":"18760db9fcdf3eb2cccb2b11f8503acb0d003d2b","subject":"[spec] X-Project-Id Pass-through","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18760db9fcdf3eb2cccb2b11f8503acb0d003d2b"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2021-10-08 08:59:29.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-01-28 05:20:34.000000000","tz":540},"subject":"OAuth2.0 Client Credentials Grant Flow Support","message":"OAuth2.0 Client Credentials Grant Flow Support\n\nThis spec proposes to allow users to optionally use an OAuth2.0 Client\nCredentials Grant flow to authorize an API client. In order to realize\nthis, we implement an OAuth2.0 authorization server as an extension of\nkeystone.\n\nImplements: blueprint oauth2-client-credentials-ext\nChange-Id: I4954c1e8f22199deb13031441c46a3565383412d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/8145886d241a48e9dd418c4b56a3a200a0ffe736"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/8145886d241a48e9dd418c4b56a3a200a0ffe736"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
