)]}'
{"id":"openstack%2Fkeystone-specs~818616","triplet_id":"openstack%2Fkeystone-specs~master~Idef5ac4083a7070f272b3e15a464a8c9dc447d47","project":"openstack/keystone-specs","branch":"master","topic":"bug/1951632","attention_set":{},"removed_from_attention_set":{"16465":{"account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"last_update":"2022-11-18 15:59:04.000000000","reason":"\u003cGERRIT_ACCOUNT_16465\u003e replied on the change","reason_account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-11-23 16:52:37.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-12-02 08:46:26.000000000","reason":"Change was submitted"},"5046":{"account":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"last_update":"2022-12-02 08:46:26.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-11-15 03:24:05.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-12-02 08:46:26.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2022-12-02 08:37:41.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-22 15:10:52.000000000","reason":"Mark Goddard replied on the change"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-12-02 08:46:26.000000000","reason":"Change was submitted"},"21420":{"account":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"last_update":"2022-01-28 16:33:57.000000000","reason":"Gage Hugo replied on the change"}},"hashtags":[],"change_id":"Idef5ac4083a7070f272b3e15a464a8c9dc447d47","subject":"Describe the need for a default service role","status":"MERGED","created":"2021-11-19 17:57:50.000000000","updated":"2022-12-02 08:47:25.000000000","submitted":"2022-12-02 08:46:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":39,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"818616-bug/1951632","meta_rev_id":"eabc35db7f772867f546b3b33daecd42eac41ed7","_number":818616,"virtual_id_number":818616,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-02 08:46:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-11-18 15:59:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2022-11-23 16:52:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2022-11-30 04:39:46.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2022-11-15 03:24:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2022-11-28 11:50:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-12-02 08:37:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2022-12-02 08:37:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-19 17:58:31.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2021-11-19 17:58:31.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2021-11-19 17:58:31.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2021-11-19 17:58:31.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2021-11-19 17:58:31.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-11-19 18:04:11.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-11-19 18:07:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-24 09:39:43.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2022-01-28 16:32:11.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-01-29 14:19:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-02-17 10:39:12.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2022-05-13 04:37:57.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REMOVED"},{"updated":"2022-10-25 16:03:01.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-11-28 11:50:00.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"8eb5d23f80f9b3041f64bac67abd8b2f979e5598","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-11-19 17:57:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa878a0e1e9c7792628faf792b6c7a2c1b6d5dcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 18:07:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8b593c4f0ef4382819d91656662b636 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcabfd253115438f98e0c295fa61e6c8 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7b94ec9ad2207bd8623712d2f9a93a6a4ee6e997","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-20 18:39:42.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f183ec0a03ae4493a64dfa7b1925b589807c70a9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-24 09:39:43.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ded8eafee74cb66c3ed731c1d1e983b9896027cc","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-01-28 16:31:42.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"de897eba4de369628f54577c573e450065136cab","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-01-28 16:32:11.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9519b2e4bee1c79e493eeee2a40a7df511b375ba","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-01-28 16:33:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"53c08ddeb6d740b26e27f95f0288b1db8da53ade","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-28 16:56:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b19861c2423a1ed0ef035dd863b654a061c9e0d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-29 02:36:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42c50ada8f42d248576222d6b3ff597f056528a9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-29 14:19:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d4a6a12a568378e4358832fd7fe53079ef483259","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-29 20:05:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb075e852a183d0425d9834264cd416b6b0b915a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-08 10:27:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e195c055677646dd669a62275f4666696c8a44e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-02-08 20:41:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41558602a77c71765638c3d9100cb5f046898541","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-09 09:15:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40418e7ef9b323236ed57e7ef7592da590945974","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-09 09:27:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b70499f5fba19b57cd950930e7f00eef38022387","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-02-20 15:34:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6ad72388f2f2b429bc645ba6527fb04463aa172","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-21 09:42:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0c5f9e33146aaabe903d0ae74436f696a73bfc3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-02-21 16:45:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"efc11a9c989a1ea7d37f5b30d8965e0dc099aff6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-21 17:01:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91de9d1b37f90f142e09c935571de2bfc445490c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-02-21 18:31:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"26b2b50f7dcbf349dcdf705824b7818688f9e6f1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 09:43:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"311d0cc274be4d060206eef0975743cc7bdb5ada","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-02-22 14:45:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c5e7f34866b7ebb8d555b144c3b1e6d7731f276","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-22 15:10:52.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"434637127be626432d235a3651d110c73175ebee","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-02-25 16:12:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9cfd3fe3c6e7fba137de386bf79ddd268e287cde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-04-01 15:12:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d3632b48922dae15904361e92235e8348f7f95fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 15:22:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59f6b714d45447729674d12429c978e9 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fadd8cff7148461e845b95a72aacc9c0 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"ac89fa210308a9705ac6cac69e1f0c61e83d355e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-04-13 09:51:16.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a2969a6ebcafbfe6d66a6a328eb8fc00c21edae0","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-05-13 04:37:57.000000000","message":"Removed reviewer Gage Hugo.","accounts_in_message":[],"_revision_number":2},{"id":"da0a9eda244d7b9e5daa2d31b1b77d92f34329cf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-10-15 21:39:06.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":2},{"id":"de30a62eec14865104fab43ff8cbe6715df4885c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-10-16 00:04:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8105192891982f72e54594810eb825349aa3a3fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-16 00:09:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d600abce0904e13906b75249ad6a1c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/780a4a96702543c998dbe0237924ffdd : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/581f3e4afd86425182457ce3c9f941d5 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":3},{"id":"69fe75f9385682acd63b5abefd1a3a354e5b4486","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-10-16 00:09:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0f1cbafcdaf676e8d3ddf0c36049a671351d19ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-16 00:15:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fc55ed5a8b2451c8a27678da858a907\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83dea397cfcd4585890a1d0387742780 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a65f48593254f528781233a54ea3b44 : SUCCESS in 2m 28s","accounts_in_message":[],"_revision_number":4},{"id":"22e1b31712489f3663eb1e42f61ceb3c85464e30","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-10-21 15:18:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f9a74bde5277d0e67b18c70089cef285020d21c7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-10-25 16:03:01.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"483f16cdc52cb2f4470b01d7fb4d09e919d3c54c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-10-25 18:17:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8c1cb678eb08b87d61475834c00de734d376480a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-10-25 18:18:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"170984e6d2dbafc557ada855b099e4494ac2161e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 18:22:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3f4c8fbd287435d9bec7d26019474b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e36aa8c77c7c4697a52243c77dee5857 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb2280f660ff4163bf37e4b545a31a7f : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":5},{"id":"a98f42a3af34a98ccfe0a60837c15bbd3dce9f9a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-10-27 13:51:47.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f8f9f247b77d248f23323642577d1e74651bc822","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-11-02 14:46:31.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"3149887b49426afe72b9edd65a6fa81dd8936e49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-02 14:52:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/331c63a3bbc74c18a914d9bb164742cd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e93cd05f077f4e739fb6f14be32f3a99 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c370ba0ea4c4999afaf6ddfdc3bc811 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":6},{"id":"1e01bc542c45a14b2369b1153c90973c5e9bafdd","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-11-09 14:33:19.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d4ea288ebc74a745238f8a08b1933daba5543bab","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-09 16:33:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7b24e16a6251af3bdedd56e7dcc08faeb89905ec","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-14 01:01:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1b894e588cb2da9e13adf500092cbd850d014d2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-14 01:04:06.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f9f1333083010ebb8be00f5364b6ecac16c10f43","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-14 01:04:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"530a4151f828a5aad55a41e19e2a1066e979ad8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 01:11:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78d57da99de14fca90c7f9de8072b6e5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a62b154143dc40a2a9eea1bb4dd3f987 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8afb783f53a3497cbd6704bae3c63bd8 : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":7},{"id":"32092365eccd8493e2f6476b9a0944c4c29e7af1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-11-15 03:24:05.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"b0c427a3e2c78cccf124e904d61a5ee90c1f3012","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-11-18 15:59:04.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0f60be060e9cfb739ac10eca61329869ce980136","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-11-23 16:52:37.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8877c79f90ded2df5b6852aa9be4749020d5238c","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-11-28 11:50:00.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"1a0630e651d18399260f854edc0d019ee241aaf2","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-11-30 04:39:46.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"473e7347cdf3123e0d1af1dbcccd64d3a57f581e","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-12-02 08:37:41.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ed10c42b2b6612f8968b079730a76077e8e57d89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 08:38:01.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"93deca20f44bfa8873a911c87dd7c7e33f3a40e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 08:46:26.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/513296e01389412a8549d838bcd238e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3040b08ddf4549128aee1e393f5b1104 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1f23e9775bc469483846b5a5a068822 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":7},{"id":"89bea7b0e762b5d3aba3f9e030c49614e2dc0ace","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 08:46:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"eabc35db7f772867f546b3b33daecd42eac41ed7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 08:47:25.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6893861e0a4e408e9a484a86ff233d92\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/a86c749c4bd141c48a7c71dc4ed0cdd2 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"75b4fb25c5b6e2c6ef4b0157c98eeb17c777c054","revisions":{"219e3d1a15f2114949654de19599f97346e33da0":{"kind":"REWORK","_number":1,"created":"2021-11-19 17:57:50.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/16/818616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/16/818616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/1"}}},"commit":{"parents":[{"commit":"18760db9fcdf3eb2cccb2b11f8503acb0d003d2b","subject":"[spec] X-Project-Id Pass-through","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/18760db9fcdf3eb2cccb2b11f8503acb0d003d2b"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@redhat.com","date":"2021-11-19 16:12:34.000000000","tz":-360},"committer":{"name":"Lance Bragstad","email":"lbragstad@redhat.com","date":"2021-11-19 17:57:14.000000000","tz":-360},"subject":"Describe the need for a default service role","message":"Describe the need for a default service role\n\nRelated-Bug: 1951632\n\nChange-Id: Idef5ac4083a7070f272b3e15a464a8c9dc447d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/219e3d1a15f2114949654de19599f97346e33da0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/219e3d1a15f2114949654de19599f97346e33da0"}]},"branch":"refs/heads/master"},"e0f27814ca6822786d1ed74aa946123cca3eea4d":{"kind":"REWORK","_number":2,"created":"2022-04-01 15:12:44.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/16/818616/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/16/818616/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/2"}}},"commit":{"parents":[{"commit":"f9f4e50737a01409391517b9b5be049419223554","subject":"Describe the need for a default manager role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f9f4e50737a01409391517b9b5be049419223554"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@redhat.com","date":"2021-11-19 16:12:34.000000000","tz":-360},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-04-01 15:12:43.000000000","tz":120},"subject":"Describe the need for a default service role","message":"Describe the need for a default service role\n\nRelated-Bug: 1951632\n\nChange-Id: Idef5ac4083a7070f272b3e15a464a8c9dc447d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/e0f27814ca6822786d1ed74aa946123cca3eea4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/e0f27814ca6822786d1ed74aa946123cca3eea4d"}]},"branch":"refs/heads/master"},"b01fcedb2c2a7cafb90d9d4655c4971b62918c1f":{"kind":"REWORK","_number":3,"created":"2022-10-16 00:04:08.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/16/818616/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/16/818616/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/3"}}},"commit":{"parents":[{"commit":"f9f4e50737a01409391517b9b5be049419223554","subject":"Describe the need for a default manager role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f9f4e50737a01409391517b9b5be049419223554"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@redhat.com","date":"2021-11-19 16:12:34.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-10-16 00:04:06.000000000","tz":-300},"subject":"Describe the need for a default service role","message":"Describe the need for a default service role\n\nRelated-Bug: 1951632\n\nChange-Id: Idef5ac4083a7070f272b3e15a464a8c9dc447d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b01fcedb2c2a7cafb90d9d4655c4971b62918c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b01fcedb2c2a7cafb90d9d4655c4971b62918c1f"}]},"branch":"refs/heads/master"},"8105af43f338274c7e39e45fc60d66195f1bfa87":{"kind":"REWORK","_number":4,"created":"2022-10-16 00:09:29.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/16/818616/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/16/818616/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/4"}}},"commit":{"parents":[{"commit":"24b290674aa84a4834130daa8057ebb0c71ba084","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/24b290674aa84a4834130daa8057ebb0c71ba084"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@redhat.com","date":"2021-11-19 16:12:34.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-10-16 00:09:23.000000000","tz":-300},"subject":"Describe the need for a default service role","message":"Describe the need for a default service role\n\nRelated-Bug: 1951632\n\nChange-Id: Idef5ac4083a7070f272b3e15a464a8c9dc447d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/8105af43f338274c7e39e45fc60d66195f1bfa87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/8105af43f338274c7e39e45fc60d66195f1bfa87"}]},"branch":"refs/heads/master"},"c522f55050ad88c4e4f7eaa3dff10c1181dc5eb3":{"kind":"REWORK","_number":5,"created":"2022-10-25 18:18:17.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/16/818616/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/16/818616/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/5"}}},"commit":{"parents":[{"commit":"7071cf3e949cd7bee726b0794b3585cfe25a7f96","subject":"remove unicode from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/7071cf3e949cd7bee726b0794b3585cfe25a7f96"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@redhat.com","date":"2021-11-19 16:12:34.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-10-25 18:18:10.000000000","tz":-300},"subject":"Describe the need for a default service role","message":"Describe the need for a default service role\n\nRelated-Bug: 1951632\n\nChange-Id: Idef5ac4083a7070f272b3e15a464a8c9dc447d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/c522f55050ad88c4e4f7eaa3dff10c1181dc5eb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/c522f55050ad88c4e4f7eaa3dff10c1181dc5eb3"}]},"branch":"refs/heads/master"},"81d836e7bf9f6a6cd0a5be6cc9a1ca3926331ee3":{"kind":"REWORK","_number":6,"created":"2022-11-02 14:46:31.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/16/818616/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/16/818616/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/6"}}},"commit":{"parents":[{"commit":"7071cf3e949cd7bee726b0794b3585cfe25a7f96","subject":"remove unicode from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/7071cf3e949cd7bee726b0794b3585cfe25a7f96"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@redhat.com","date":"2021-11-19 16:12:34.000000000","tz":-360},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2022-11-02 14:46:31.000000000","tz":0},"subject":"Describe the need for a default service role","message":"Describe the need for a default service role\n\nRelated-Bug: 1951632\n\nChange-Id: Idef5ac4083a7070f272b3e15a464a8c9dc447d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/81d836e7bf9f6a6cd0a5be6cc9a1ca3926331ee3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/81d836e7bf9f6a6cd0a5be6cc9a1ca3926331ee3"}]},"branch":"refs/heads/master"},"75b4fb25c5b6e2c6ef4b0157c98eeb17c777c054":{"kind":"REWORK","_number":7,"created":"2022-11-14 01:04:06.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/16/818616/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/16/818616/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/16/818616/7"}}},"commit":{"parents":[{"commit":"7071cf3e949cd7bee726b0794b3585cfe25a7f96","subject":"remove unicode from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/7071cf3e949cd7bee726b0794b3585cfe25a7f96"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@redhat.com","date":"2021-11-19 16:12:34.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-11-14 01:03:58.000000000","tz":-360},"subject":"Describe the need for a default service role","message":"Describe the need for a default service role\n\nRelated-Bug: 1951632\n\nChange-Id: Idef5ac4083a7070f272b3e15a464a8c9dc447d47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/75b4fb25c5b6e2c6ef4b0157c98eeb17c777c054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/75b4fb25c5b6e2c6ef4b0157c98eeb17c777c054"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[]}
