)]}'
{"id":"openstack%2Fkeystone-specs~839909","triplet_id":"openstack%2Fkeystone-specs~master~Iddc30b9521b61d9083c2b1f6e8a6707196ea0a57","project":"openstack/keystone-specs","branch":"master","topic":"setuptools-issue-3197","attention_set":{},"removed_from_attention_set":{"16465":{"account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"last_update":"2022-06-07 13:52:54.000000000","reason":"Kristi Nikolla replied on the change"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-06-07 14:02:01.000000000","reason":"Change was submitted"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-06-07 14:02:01.000000000","reason":"Change was submitted"},"13478":{"account":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"last_update":"2022-06-07 14:02:01.000000000","reason":"Change was submitted"},"5046":{"account":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"last_update":"2022-06-07 14:02:01.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2022-06-07 14:02:01.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2022-06-07 14:02:01.000000000","reason":"Change was submitted"},"33455":{"account":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"last_update":"2022-06-07 14:02:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iddc30b9521b61d9083c2b1f6e8a6707196ea0a57","subject":"Disable auto-discovery for setuptools and update python testing","status":"MERGED","created":"2022-04-29 11:57:37.000000000","updated":"2022-06-07 14:04:48.000000000","submitted":"2022-06-07 14:02:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839909-setuptools-issue-3197","meta_rev_id":"627b9a58c4d4b0aab824f5c4382058e8efe930de","_number":839909,"virtual_id_number":839909,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-07 14:02:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-06-07 13:52:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2022-06-07 13:52:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 12:09:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-04 12:58:48.000000000","updated_by":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"reviewer":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"state":"REVIEWER"},{"updated":"2022-05-31 17:17:05.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"state":"REVIEWER"},{"updated":"2022-05-31 17:17:05.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2022-05-31 17:17:05.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2022-05-31 17:17:05.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-05-31 17:17:05.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-05-31 17:17:05.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"}],"messages":[{"id":"bf25dacb91c126382ec9274206286175301efe8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-04-29 11:57:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4376fb9c7c433bf6439f97dccd242feee0f8000","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 12:09:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92d014a2b6b24597ab79563917c97f70 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e09a7fa0759043b6b3c38470c715bf4b : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46e54a52bd234c8287bcaa4ba7c2d09f : SUCCESS in 3m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff9280d736bb48738d3039039f87d282 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c82c04b0d09482aa2655ac95a3007ae : SUCCESS in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5fa89d6157f2feb06014ef34bb39a650c906509f","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2022-05-04 12:58:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"57d2051741031bcef79f1f469bafc68e470d25c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-05-31 17:04:39.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c484c07cbfd3619692ee2e94837335ed9868005d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 17:14:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2edf9568343f487f972752cdc8c28a4e : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2dccc8da3a043a5aec34cb56c995a9e : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1eb0e757aeaf4908881548d33d1036b3 : FAILURE in 2m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0deeca443cb5486bbb344c8ca7e5bb8a : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/851f8bfd1b894209952298de974d6b62 : SUCCESS in 3m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"997315b8d0ea90c99d39889bb0c0c5a72ad2b7b4","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-05-31 17:17:05.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d53da5bebb7ba075c9d2debf3b507c37177240e2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-01 13:56:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1802fc9805759310c3db0364d9d4fe04b85cf3d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2022-06-07 10:19:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"172584e75c8cc979b9eedbd9a9ac20b384b17f2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 10:28:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0ef0d9a20594aec88456b8b63e4d598 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75655f914c084cc18f6da943b1d23d4c : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5241559e8ca4c149e921c86120c959a : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3dae36f1685a44daae4d4dbb35563162 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db256ac405a84c21b1043d78c69b0ed4 : SUCCESS in 3m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"282c083b66f5c699a72f3cde03f5a10139104efb","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-06-07 13:52:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b4671fe234feecd86032078a89bd3f2fe208fc77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 13:53:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"85735b7e7ba2f793f38341f383ab93b3686ab938","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 14:02:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dc52ae83ccc4529b30c276664960c96 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbdfb2816c9846518167a3749953dade : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b63030750c54c968fe117adcd45ff5e : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86cb536b510f4d3aa52984b488bf8426 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":3},{"id":"880e9f9b2d8a2d72e533ffafbb9d0f6867e92cb6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 14:02:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"627b9a58c4d4b0aab824f5c4382058e8efe930de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 14:04:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/88b0384f73794647b1e15d44f4649bd8 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6df4f46055c64e877ce555123b5be7ab395429cc","revisions":{"bee7a65573666a154fc42639fdc736e12676a6e5":{"kind":"REWORK","_number":1,"created":"2022-04-29 11:57:37.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/09/839909/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/09/839909/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/09/839909/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/09/839909/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/09/839909/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/09/839909/1"}}},"commit":{"parents":[{"commit":"f9f4e50737a01409391517b9b5be049419223554","subject":"Describe the need for a default manager role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f9f4e50737a01409391517b9b5be049419223554"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-04-29 11:55:59.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-04-29 11:57:35.000000000","tz":120},"subject":"Disable auto-discovery for setuptools","message":"Disable auto-discovery for setuptools\n\nWith setuptools release 61.0.0 docs build started to fail:\nerror: Multiple top-level packages discovered in a flat-layout:\n[\u0027specs\u0027, \u0027attic\u0027, \u0027superseded\u0027].\n\nThis bug is mentioned in setuptools issue 3197 [0], and the suggested\nworkaround is to disable auto-discovery by adding \u0027py_modules\u003d[]\u0027 in\nsetup.py.\n\n[0] https://github.com/pypa/setuptools/issues/3197\n\nChange-Id: Iddc30b9521b61d9083c2b1f6e8a6707196ea0a57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/bee7a65573666a154fc42639fdc736e12676a6e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/bee7a65573666a154fc42639fdc736e12676a6e5"}]},"branch":"refs/heads/master"},"e517646709f671148f331aae928d2b2ac2aed126":{"kind":"NO_CHANGE","_number":2,"created":"2022-05-31 17:04:39.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/09/839909/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/09/839909/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/09/839909/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/09/839909/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/09/839909/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/09/839909/2"}}},"commit":{"parents":[{"commit":"f9f4e50737a01409391517b9b5be049419223554","subject":"Describe the need for a default manager role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f9f4e50737a01409391517b9b5be049419223554"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-04-29 11:55:59.000000000","tz":120},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-31 17:04:25.000000000","tz":540},"subject":"Disable auto-discovery for setuptools","message":"Disable auto-discovery for setuptools\n\nWith setuptools release 61.0.0 docs build started to fail:\nerror: Multiple top-level packages discovered in a flat-layout:\n[\u0027specs\u0027, \u0027attic\u0027, \u0027superseded\u0027].\n\nThis bug is mentioned in setuptools issue 3197 [0], and the suggested\nworkaround is to disable auto-discovery by adding \u0027py_modules\u003d[]\u0027 in\nsetup.py.\n\n[0] https://github.com/pypa/setuptools/issues/3197\n\nChange-Id: Iddc30b9521b61d9083c2b1f6e8a6707196ea0a57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/e517646709f671148f331aae928d2b2ac2aed126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/e517646709f671148f331aae928d2b2ac2aed126"}]},"branch":"refs/heads/master"},"6df4f46055c64e877ce555123b5be7ab395429cc":{"kind":"REWORK","_number":3,"created":"2022-06-07 10:19:00.000000000","uploader":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"ref":"refs/changes/09/839909/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/09/839909/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/09/839909/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/09/839909/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/09/839909/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/09/839909/3"}}},"commit":{"parents":[{"commit":"f9f4e50737a01409391517b9b5be049419223554","subject":"Describe the need for a default manager role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f9f4e50737a01409391517b9b5be049419223554"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-04-29 11:55:59.000000000","tz":120},"committer":{"name":"Boris Bobrov","email":"b.bobrov@sap.com","date":"2022-06-07 10:16:12.000000000","tz":120},"subject":"Disable auto-discovery for setuptools and update python testing","message":"Disable auto-discovery for setuptools and update python testing\n\nWith setuptools release 61.0.0 docs build started to fail:\nerror: Multiple top-level packages discovered in a flat-layout:\n[\u0027specs\u0027, \u0027attic\u0027, \u0027superseded\u0027].\n\nThis bug is mentioned in setuptools issue 3197 [0], and the suggested\nworkaround is to disable auto-discovery by adding \u0027py_modules\u003d[]\u0027 in\nsetup.py.\n\nAlso use recent python versions because the old ones are no longer available.\n\nThese 2 unrelated changes need to be merged together in order to unblock\nthe gate.\n\n[0] https://github.com/pypa/setuptools/issues/3197\n\nChange-Id: Iddc30b9521b61d9083c2b1f6e8a6707196ea0a57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6df4f46055c64e877ce555123b5be7ab395429cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6df4f46055c64e877ce555123b5be7ab395429cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
