)]}'
{"id":"openstack%2Fkeystone-specs~843765","triplet_id":"openstack%2Fkeystone-specs~master~I67e030c183631bd421cc93ceb767f60fa178238a","project":"openstack/keystone-specs","branch":"master","topic":"bp/support-oauth2-mtls","attention_set":{},"removed_from_attention_set":{"33920":{"account":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"last_update":"2022-12-20 14:08:03.000000000","reason":"Change was submitted"},"16465":{"account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"last_update":"2022-12-20 14:02:18.000000000","reason":"\u003cGERRIT_ACCOUNT_16465\u003e replied on the change","reason_account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},"25701":{"account":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"last_update":"2022-12-20 14:08:03.000000000","reason":"Change was submitted"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-09-16 17:55:54.000000000","reason":"\u003cGERRIT_ACCOUNT_7973\u003e replied on the change","reason_account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},"5046":{"account":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"last_update":"2022-12-20 14:08:03.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2022-07-01 15:47:29.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"34712":{"account":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"last_update":"2022-12-20 14:08:03.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2022-08-02 13:27:30.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},"33455":{"account":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"last_update":"2022-12-20 14:08:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I67e030c183631bd421cc93ceb767f60fa178238a","subject":"OAuth 2.0 Mutual-TLS Support","status":"MERGED","created":"2022-05-30 05:08:53.000000000","updated":"2022-12-20 14:09:11.000000000","submitted":"2022-12-20 14:08:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":49,"unresolved_comment_count":14,"has_review_started":true,"submission_id":"843765-bp/support-oauth2-mtls","meta_rev_id":"a65a38824711431a2313a0784620f4fc9e7b3010","_number":843765,"virtual_id_number":843765,"owner":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-20 14:08:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-12-20 14:02:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2022-12-16 15:12:10.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2022-12-16 16:38:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2022-12-20 14:03:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"}],"reviewers":{"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}],"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-30 05:13:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-31 14:40:45.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"state":"REVIEWER"},{"updated":"2022-05-31 14:40:45.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2022-05-31 14:40:45.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-05-31 14:40:45.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2022-05-31 14:40:45.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-05-31 14:40:45.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2022-06-03 05:22:19.000000000","updated_by":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"reviewer":{"_account_id":25701,"name":"Yasufumi Ogawa","email":"yasufum.o@gmail.com","username":"yasufum"},"state":"REVIEWER"},{"updated":"2022-07-05 02:08:42.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":34712,"name":"Yuta Kazato","display_name":"Yuta Kazato","email":"yuta.kazato.nw@hco.ntt.co.jp","username":"kazatoy-ntt"},"state":"REVIEWER"},{"updated":"2022-12-16 15:37:05.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"}],"messages":[{"id":"763532d3cb56b639cb80b6e010b6466d7d912d58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-05-30 05:08:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a024af1bf5a91788fa770a03e1a89101cdc1469a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 05:13:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83b9a769e68d42dd8a7572a6df3db6ab : FAILURE in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56fe7e5c4dc74aab8af58878e77241e0 : FAILURE in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"9782b6e7477bd8a0772b291af4b2ffb9404f2745","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-05-30 06:35:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cbd9f5c0a6256ba4bd951c9e8557b331cebeb95e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 06:40:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/884a6b26f71b448ab2679be1e0e273aa : FAILURE in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af512f37cefc4dd2a182b3a4972dfebe : FAILURE in 2m 33s","accounts_in_message":[],"_revision_number":1},{"id":"abb00f78a8218d1a885e0b6dc8bb12e42937ec25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-05-31 10:04:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ebc29bf56f3909c5d8e70fa3fa4d4ee4653f2abf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 10:09:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd8502772dd14609b105bfc33a9a8422 : FAILURE in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51e7b72e54804972a504b6fd120b35e8 : FAILURE in 2m 34s","accounts_in_message":[],"_revision_number":2},{"id":"2b7cd02020914a7e2d4365febd5a5578d69d00f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-05-31 10:14:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"228e459f5819d1263bf37bb2eab66bc35068863b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 10:24:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2907bf8ecb241e3a7c8f0486ca45893 : FAILURE in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/637e23c33c3c42e9a734a3bfac0a0029 : FAILURE in 2m 57s","accounts_in_message":[],"_revision_number":3},{"id":"09c0796b8ceb61d2635429d933098619f70d3156","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-05-31 14:59:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"95582de6c6212195c058fd232676900625a466de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 15:13:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/536835c673d94aa2a1ef3f3e49162cd7 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/626f0e41c96b42bebfbfb54c0d94d766 : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef4906d252d745858cc9c3be7f1a91f6 : FAILURE in 2m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c0b7bb620764f64a4d118363ad8b91c : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a55d4da1cbcd41558dd016b1eaf157fe : SUCCESS in 2m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f1f9baa1b32b43e55598350ff4552d93ca5c85ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-05-31 17:04:39.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"8d7773fd7850b3c15f521f1bc14cc3a116a602ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 17:14:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c63b498db7ba4dffb17701055719d8f4 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f763fcf721ec4ac8b31ff823eaa5839f : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8f3837fbcb24615b2a60e81ea213134 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48e2122df97f44dbbdda3587ca230df1 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5dd93ba37eba453785a38169bbf0d30f : SUCCESS in 3m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"00b7b09f0ae8e2ac02771d80af06bed6a0b36b4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-06-29 10:15:26.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"de3a256eba6dca3127fa3a47445dac00191beaee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 10:24:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ac2bb7ce8af4cada2c87dbb0f9b47dc : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f093b2820cff484b9e730b2ef39d493d : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e35bc2f22dee420aa054d1071787d7a6 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a94951962364103b2fd6dbecac218ec : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2758ba392974fcfaf3ee4d389cc0c22 : SUCCESS in 3m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3f0bce32150b5e4008fc64c5a6283a334f959101","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-06-30 06:00:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"bd286308c8aecd07ab6bfddd03a67d48b974381c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 06:07:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4726067bbdb6437797702d1348c1994a : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b65f3b3970db40d3837bbda777d0ad44 : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5496d9f19e974f59853bb63e90e990ab : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03f1b5d6ae6f4139aa0df2a4d4aa692a : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95080a5a3bba4f4da6cef152cb853126 : SUCCESS in 3m 08s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"2b956926d5b908235522118a57ccec0c8fe64702","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-07-01 15:47:29.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"ace621875131ac91a12439da4e5eada9f9b2f802","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-07-07 11:39:19.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d5b889b5bcb71846eb93b0ac26c901a914d9cadc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-07-07 11:43:05.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"15788a251e0b109c356ff84f058cc7e2bcd0a54b","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-07-07 11:43:55.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"98857eb5a8a551e8f6bad3a8746e2cb61a7f5fe3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 11:56:05.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e40ef7445114d028de155b87db25422\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51d68fa866334a17a8fe6fc4aac12a77 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/001730e045e04ed7917796e24f3a1bd1 : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2537127c1e6a442f8e2f7d11f29214d8 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d87bfb65cdad4ef497b10f6f875e2f2e : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05996855c0fa4d99acdcdcbbb9ff9949 : SUCCESS in 2m 43s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f4af7ad8e3cab3ebd4170f0e979a68403ee6728a","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-07-08 15:12:35.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"066f14e8ff3e689f4c297efe2dc99e4aba8fc6e0","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-07-11 03:38:13.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a63e8c8e65eb204043043e454445191c1cf0b4db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-07-11 07:47:14.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"6061319fe806508e59f42ea5d4e214223fdd4806","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-07-11 07:51:45.000000000","message":"Patch Set 10:\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"e5cd6095dd5259cbeb6fa76414ff6fc591191cfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 07:52:15.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00e3fca02617482494286eb7a33c6208\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b37c19147c324aa0ad41865a8bacc138 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2d3909404f745eda43dc4e28a6245a8 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a45c24a6fec4e0caf6520f0e3de4194 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04a5f7982e0243db933c6a0d7a0e62f0 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a22af500a7146dc91fdb742d6a4ec08 : SUCCESS in 2m 59s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1ac3d68798923eb8584efd3df722ac5c6dcb656d","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-07-15 10:34:30.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"b3164622630963d444c12a021b088c29e2de3dd3","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-07-22 14:42:27.000000000","message":"Patch Set 10:\n\n(10 comments)","accounts_in_message":[],"_revision_number":10},{"id":"15527d3b4ec0f72f570d82b163177264f0d774d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-07-27 08:16:38.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"e9ae6e6eefc04c2b87c31a110db4b68952693f58","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-07-27 08:19:40.000000000","message":"Patch Set 10:\n\n(12 comments)","accounts_in_message":[],"_revision_number":10},{"id":"54211b234395b968af4830e15685596df4691155","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 08:21:40.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78ed41097e8e4b2eb34108681f3378ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90a09b9594564ed4aa6a6f4bb7d2a377 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ed26d5a60854d65b854af3aad16343f : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e643e30136cb4332829ad839545ae52b : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/769b0d2fc4d54fc5ac001222c3175f87 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9452783909eb42a89bb179b9fdd3846e : SUCCESS in 4m 06s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b4cf22640646a86f0ec2cea8b41287b56c894246","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-07-27 08:39:37.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"c9d49943387ebdbecbd15d206d22ae727a4d7aa3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 08:44:29.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44aee542483240abb1d7b2361fd05eef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2b6227eaa954158abc00d242b62bfd5 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daffc1c9a63646f9a4428c4f7e946221 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b30b9fce573422ea547933995400a34 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aec84013998348ed9c4768067a3de887 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f20d58f6d794d41a073c496996d467d : SUCCESS in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c6d3f477f3a2db55b8b862aa6338ccacba0e315f","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-07-27 08:45:44.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"0e8c47f38ce2c302247f5a628f33e6ee54812080","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-08-02 13:27:30.000000000","message":"Patch Set 12: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"87b5791ac6f272a716cdde7de2323e22decbd8d7","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-08-05 11:27:05.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"f8a0a415223152e315f4c7c1cdbfb118100e5935","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-08-05 15:26:07.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c70a76801d8fbb7c1c188db820ef24a15fc16d60","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-08-09 14:50:26.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"285e531d3b1755c17c8616629b8f36951a84bd21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-08-10 09:49:38.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"f739b34b1de1da1f803eebd3644816753eb71974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 10:11:43.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c78189fe4e60443893ef024adaab100d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a50a2cd5f594696b32442bd8e61aace : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c523833927fe4ae7b9203c74955ee769 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f017477475c14efab311ede4b7ca86a3 : SUCCESS in 12m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7758cc24b233417ba85b1afc96b7067a : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cad4db4ddd91461ca25c7af3f1e6c843 : SUCCESS in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"0ace8d370c9a97bb018ed0a82cbe6aade05463e3","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-08-10 12:15:46.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"dfeba11c04bf4244deb9d6886622306e479de956","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-08-26 12:25:18.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"7258722149be6cdde79920d69503b0191fff39c7","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-08-26 12:30:47.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d756fe9edc38167f1c9e16149122e918f28f261b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 12:41:26.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69fcf86ec6c24f9fb5815d84a37c5ee7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08a68698dc0f4a359595091938b404b2 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b2d6995adf1444abad28183dcb7e207 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac40670d7bca4443a9f66a7aaee27fc6 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46d9345c18cb4132b10db8d65c26d2db : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33e98102c410452baf58b4f8749ece2d : SUCCESS in 4m 01s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"410ec7db0eea7be4683fac919c4b49fb00c96c07","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-09-16 17:55:54.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"874d5f32b9d2448410b75e32ae490c2f2bc03731","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-09-20 15:44:14.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"baae954ddf1e6e1c8afff08c1e3d2ce4538cfc75","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-09-20 15:46:08.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"c7cc38ccc75591061d8647e6c06acc6c871fe1a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 15:49:45.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d82a6794c8f4fd4831f061d678445c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e8c38c6f8a74e93b9123cf90ba6df8f : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcfcc273c2dd450182c2741640d4a565 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/581d5ca6cbe04590af3c3873e6b925e4 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8201b590b20b4e75b33827c915aa3e97 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a3012374a884333b1993752a1d34a36 : SUCCESS in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"1a562c360bf3590a247052ff1107a338e9464958","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-10-17 03:13:56.000000000","message":"Topic set to bp/support-oauth2-mtls","accounts_in_message":[],"_revision_number":15},{"id":"bbbc103e6d36cdb58ac632a275609eac763ef571","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-10-17 03:47:35.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"8ca26fe86f888c5182e877c62f0543ca62fb2796","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-10-17 03:52:15.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"5836337987545499c78b17b4a311f6c5fc9e6750","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 03:57:36.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e11690999457427dbf860605daae16a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b37e279c83de405c8106db44765a9838 : SUCCESS in 3m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b43bc30dae3144639ab090f0da51a035 : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8feed4407324ada8cf0eebc2e3c7d64 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c03744f61eb4f84a29a569e1f1f3c5e : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/914ce340bf7a417094f2dfca28d2e110 : SUCCESS in 3m 08s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"548f504363f8168b379c8bc53d9cc9411bb7e9f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-12-02 13:03:37.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"6f7b940942d4bfe2da05a6d6a47e400b724052b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-02 13:13:26.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7044b53256a94f78acd8d1829f9f6796\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a7fb99c0211426e817bace22392baa6 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2875c46a76304b0184eae3382c412f64 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0df2d4196d84dc89c9879a92a5501fc : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b64159ffb948474d8e85392cb7ba9acd : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad6cfbb10a8645fdadbd58de7dbac007 : SUCCESS in 2m 59s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"8ffd6af5561d53d73326e5d3943ccc1e1bfe7318","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-12-09 14:42:14.000000000","message":"Patch Set 17: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"79fdbd49c97c719bd65b37003a75d4254b63b968","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-12-13 14:54:22.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"b3219f396a3de3bf2c72531cce81d351ea294c7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 14:59:26.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0ae8bede00d4ce1af98b0da3783e006\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9bf17f1f2e44ea3a587995191fb14da : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e6373d76f134ba7a061916918ab0653 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa259898897941d183ecf73b95290ded : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b6c51981f664ffe9e0dc18c19a3a80f : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8ee842e2cb34b9db21c02243ebb6b7a : SUCCESS in 3m 09s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"53ba4bb71c940fa0f86b9e808cb6fe0bbe1d4a63","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-12-13 14:59:59.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"980615654318e10f4057333b89926456da03686d","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-12-16 15:12:10.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"07d02cb0dc73f1528dbb95b9dabdabab8d292cd1","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2022-12-16 15:37:05.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"499a967fa0bc864fbf33e3f13e667524fdbf032b","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-12-16 16:38:49.000000000","message":"Patch Set 18: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"f54f4893571dcd18903cfd9ddb5824755b96fd43","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-12-20 14:02:18.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"bfb6d51746664b337bb32d7d3db510a5cb948e26","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-12-20 14:03:00.000000000","message":"Patch Set 18: Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"e6b8ba46e5e1643469ad29cee5df78007075219f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 14:03:27.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"2559f26b57afc3451451a39534399cada5031dc9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 14:08:03.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9eaf42a63e04721ba68dd3286b3eabf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9a559d7194341f092e3382feae10c82 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58f8d84ffef64a3f87728dc8856e6b03 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe901f3a38ab474d8091ed5cd8e6545c : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2a47311f9644a1893c74b51aa2d2648 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":18},{"id":"2d7bbd5eed42f464634af4bc8dc12eff017bd9ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 14:08:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":18},{"id":"a65a38824711431a2313a0784620f4fc9e7b3010","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-20 14:09:11.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8b23458598640e78c7967b69bff733b\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/ca2350d9519d4920b4ff3719db6d136d : SUCCESS in 44s","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"12f37d354808921834b4beb5193d95adae4aa3ad","revisions":{"f8dc690cbcfe64798d6bff101de3472684833e0f":{"kind":"REWORK","_number":1,"created":"2022-05-30 05:08:53.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/1"}}},"commit":{"parents":[{"commit":"f9f4e50737a01409391517b9b5be049419223554","subject":"Describe the need for a default manager role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f9f4e50737a01409391517b9b5be049419223554"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 05:08:50.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f8dc690cbcfe64798d6bff101de3472684833e0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f8dc690cbcfe64798d6bff101de3472684833e0f"}]},"branch":"refs/heads/master"},"07daf8dfac66ead3e0ebfa29629f1b268759a983":{"kind":"REWORK","_number":2,"created":"2022-05-31 10:04:15.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/2"}}},"commit":{"parents":[{"commit":"f9f4e50737a01409391517b9b5be049419223554","subject":"Describe the need for a default manager role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f9f4e50737a01409391517b9b5be049419223554"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-31 10:03:58.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/07daf8dfac66ead3e0ebfa29629f1b268759a983"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/07daf8dfac66ead3e0ebfa29629f1b268759a983"}]},"branch":"refs/heads/master"},"4272dff8190e6f92ecc075d71a6b55a4825c4867":{"kind":"REWORK","_number":3,"created":"2022-05-31 10:14:50.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/3"}}},"commit":{"parents":[{"commit":"f9f4e50737a01409391517b9b5be049419223554","subject":"Describe the need for a default manager role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f9f4e50737a01409391517b9b5be049419223554"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-31 10:14:40.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/4272dff8190e6f92ecc075d71a6b55a4825c4867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/4272dff8190e6f92ecc075d71a6b55a4825c4867"}]},"branch":"refs/heads/master"},"6c166e6f2348bd86d670feedfa921dd9f4f6f1c1":{"kind":"REWORK","_number":4,"created":"2022-05-31 14:59:18.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/4"}}},"commit":{"parents":[{"commit":"f9f4e50737a01409391517b9b5be049419223554","subject":"Describe the need for a default manager role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f9f4e50737a01409391517b9b5be049419223554"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-31 14:59:11.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6c166e6f2348bd86d670feedfa921dd9f4f6f1c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6c166e6f2348bd86d670feedfa921dd9f4f6f1c1"}]},"branch":"refs/heads/master"},"c7ad86f82d328db77ab18191c4e233f79c54f574":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-05-31 17:04:39.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/5"}}},"commit":{"parents":[{"commit":"6db0ee51a4a450d78a9098ffdea4972fcdc572a1","subject":"Update python testing template to latest version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6db0ee51a4a450d78a9098ffdea4972fcdc572a1"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-31 17:04:25.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/c7ad86f82d328db77ab18191c4e233f79c54f574"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/c7ad86f82d328db77ab18191c4e233f79c54f574"}]},"branch":"refs/heads/master"},"b8effbcfa31831ab6c86c8730366af9b31112a25":{"kind":"REWORK","_number":6,"created":"2022-06-29 10:15:26.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/6"}}},"commit":{"parents":[{"commit":"6df4f46055c64e877ce555123b5be7ab395429cc","subject":"Disable auto-discovery for setuptools and update python testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6df4f46055c64e877ce555123b5be7ab395429cc"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-06-29 10:15:16.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b8effbcfa31831ab6c86c8730366af9b31112a25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b8effbcfa31831ab6c86c8730366af9b31112a25"}]},"branch":"refs/heads/master"},"2f3299097497987ac64b2648fc86ce7f213017e1":{"kind":"REWORK","_number":7,"created":"2022-06-30 06:00:24.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/7"}}},"commit":{"parents":[{"commit":"6df4f46055c64e877ce555123b5be7ab395429cc","subject":"Disable auto-discovery for setuptools and update python testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6df4f46055c64e877ce555123b5be7ab395429cc"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-06-30 06:00:15.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/2f3299097497987ac64b2648fc86ce7f213017e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/2f3299097497987ac64b2648fc86ce7f213017e1"}]},"branch":"refs/heads/master"},"59a93f11cdbdc0edce458bb8fb1d810a727f461a":{"kind":"REWORK","_number":8,"created":"2022-07-07 11:39:19.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/8"}}},"commit":{"parents":[{"commit":"6df4f46055c64e877ce555123b5be7ab395429cc","subject":"Disable auto-discovery for setuptools and update python testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6df4f46055c64e877ce555123b5be7ab395429cc"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-07-07 11:38:42.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/59a93f11cdbdc0edce458bb8fb1d810a727f461a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/59a93f11cdbdc0edce458bb8fb1d810a727f461a"}]},"branch":"refs/heads/master"},"a9e0b95c8baf548e1d45ce5a63ff1fedd16915fb":{"kind":"REWORK","_number":9,"created":"2022-07-07 11:43:05.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/9"}}},"commit":{"parents":[{"commit":"6df4f46055c64e877ce555123b5be7ab395429cc","subject":"Disable auto-discovery for setuptools and update python testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6df4f46055c64e877ce555123b5be7ab395429cc"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-07-07 11:42:52.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/a9e0b95c8baf548e1d45ce5a63ff1fedd16915fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/a9e0b95c8baf548e1d45ce5a63ff1fedd16915fb"}]},"branch":"refs/heads/master"},"2f8cec5e3718e94baae8aeeac6aa67a20c250acc":{"kind":"REWORK","_number":10,"created":"2022-07-11 07:47:14.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/10"}}},"commit":{"parents":[{"commit":"6df4f46055c64e877ce555123b5be7ab395429cc","subject":"Disable auto-discovery for setuptools and update python testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6df4f46055c64e877ce555123b5be7ab395429cc"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-07-11 07:47:05.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/2f8cec5e3718e94baae8aeeac6aa67a20c250acc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/2f8cec5e3718e94baae8aeeac6aa67a20c250acc"}]},"branch":"refs/heads/master"},"d1767e8207660a56283df31f1b2621cf2879473c":{"kind":"REWORK","_number":11,"created":"2022-07-27 08:16:38.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/11"}}},"commit":{"parents":[{"commit":"6df4f46055c64e877ce555123b5be7ab395429cc","subject":"Disable auto-discovery for setuptools and update python testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6df4f46055c64e877ce555123b5be7ab395429cc"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-07-27 08:16:27.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/d1767e8207660a56283df31f1b2621cf2879473c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/d1767e8207660a56283df31f1b2621cf2879473c"}]},"branch":"refs/heads/master"},"f8f43587b58f67c5ca673be79056fb4641ecf784":{"kind":"REWORK","_number":12,"created":"2022-07-27 08:39:37.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/12"}}},"commit":{"parents":[{"commit":"6df4f46055c64e877ce555123b5be7ab395429cc","subject":"Disable auto-discovery for setuptools and update python testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6df4f46055c64e877ce555123b5be7ab395429cc"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-07-27 08:39:28.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f8f43587b58f67c5ca673be79056fb4641ecf784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f8f43587b58f67c5ca673be79056fb4641ecf784"}]},"branch":"refs/heads/master"},"84154fdc99cc236667c4599333ead385db0b13a2":{"kind":"REWORK","_number":13,"created":"2022-08-10 09:49:38.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/13"}}},"commit":{"parents":[{"commit":"6df4f46055c64e877ce555123b5be7ab395429cc","subject":"Disable auto-discovery for setuptools and update python testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6df4f46055c64e877ce555123b5be7ab395429cc"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-08-10 09:49:25.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/84154fdc99cc236667c4599333ead385db0b13a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/84154fdc99cc236667c4599333ead385db0b13a2"}]},"branch":"refs/heads/master"},"6d377f89dfc8f5a8e0860e80cd290c84db20aabc":{"kind":"REWORK","_number":14,"created":"2022-08-26 12:25:18.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/14"}}},"commit":{"parents":[{"commit":"6df4f46055c64e877ce555123b5be7ab395429cc","subject":"Disable auto-discovery for setuptools and update python testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6df4f46055c64e877ce555123b5be7ab395429cc"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-08-26 12:24:54.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6d377f89dfc8f5a8e0860e80cd290c84db20aabc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6d377f89dfc8f5a8e0860e80cd290c84db20aabc"}]},"branch":"refs/heads/master"},"2d833b6eedbfe908cfad46f24a1958a3e344cdb6":{"kind":"REWORK","_number":15,"created":"2022-09-20 15:44:14.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/15"}}},"commit":{"parents":[{"commit":"6df4f46055c64e877ce555123b5be7ab395429cc","subject":"Disable auto-discovery for setuptools and update python testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6df4f46055c64e877ce555123b5be7ab395429cc"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-09-20 15:44:05.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/2d833b6eedbfe908cfad46f24a1958a3e344cdb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/2d833b6eedbfe908cfad46f24a1958a3e344cdb6"}]},"branch":"refs/heads/master"},"ca99a20a8ebafc73c3f3d709599b2a5f4d47246b":{"kind":"REWORK","_number":16,"created":"2022-10-17 03:47:35.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/16"}}},"commit":{"parents":[{"commit":"6df4f46055c64e877ce555123b5be7ab395429cc","subject":"Disable auto-discovery for setuptools and update python testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/6df4f46055c64e877ce555123b5be7ab395429cc"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-10-17 03:47:12.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/ca99a20a8ebafc73c3f3d709599b2a5f4d47246b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/ca99a20a8ebafc73c3f3d709599b2a5f4d47246b"}]},"branch":"refs/heads/master"},"b4c87b220277c8e7c8cd5f64a5a10df832a5fdae":{"kind":"REWORK","_number":17,"created":"2022-12-02 13:03:37.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/17"}}},"commit":{"parents":[{"commit":"75b4fb25c5b6e2c6ef4b0157c98eeb17c777c054","subject":"Describe the need for a default service role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/75b4fb25c5b6e2c6ef4b0157c98eeb17c777c054"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-12-02 13:02:59.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b4c87b220277c8e7c8cd5f64a5a10df832a5fdae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b4c87b220277c8e7c8cd5f64a5a10df832a5fdae"}]},"branch":"refs/heads/master"},"12f37d354808921834b4beb5193d95adae4aa3ad":{"kind":"REWORK","_number":18,"created":"2022-12-13 14:54:22.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/65/843765/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/65/843765/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/65/843765/18"}}},"commit":{"parents":[{"commit":"75b4fb25c5b6e2c6ef4b0157c98eeb17c777c054","subject":"Describe the need for a default service role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/75b4fb25c5b6e2c6ef4b0157c98eeb17c777c054"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-30 02:41:30.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-12-13 14:54:11.000000000","tz":540},"subject":"OAuth 2.0 Mutual-TLS Support","message":"OAuth 2.0 Mutual-TLS Support\n\nThis spec proposes to Provide the option for users to\nproof-of-possession of OAuth2.0 access token based on RFC8705 OAuth 2.0\nMutual-TLS Client Authentication and Certificate-Bound Access Tokens.\nUsers will be able to authenticate their OAuth2.0 client with a client\ncertificate instead of using Basic authentication with\nclient_id/client_secret to prevent a token from being used by a\nmalicious client. This protects Keystone Identity and other OpenStack\nservices from spoofed OAuth clients.\n\nChange-Id: I67e030c183631bd421cc93ceb767f60fa178238a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/12f37d354808921834b4beb5193d95adae4aa3ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/12f37d354808921834b4beb5193d95adae4aa3ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[]}
