)]}'
{"id":"openstack%2Fkeystone-specs~844100","triplet_id":"openstack%2Fkeystone-specs~master~Idcfe2957fa83723f6de927ede36d21c7b5e6431e","project":"openstack/keystone-specs","branch":"master","topic":"bug/1976387","attention_set":{},"removed_from_attention_set":{"33455":{"account":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"last_update":"2022-05-31 17:06:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idcfe2957fa83723f6de927ede36d21c7b5e6431e","subject":"[WIP] Fix document build","status":"ABANDONED","created":"2022-05-31 15:49:05.000000000","updated":"2022-05-31 17:06:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b6854da731565252dbd3d32648fd358343790a2a","_number":844100,"virtual_id_number":844100,"owner":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-31 15:58:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"da02f0b91a99cb85aa3caddb4d1621dd6a724045","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-05-31 15:49:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd4c49b8e7c4c81ae632f4b51f71a36175103490","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 15:58:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d9d27c7c0344fcf84615867da6922a5 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cb227aaec0e47c9807b9e8ecd34d3c5 : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1c79c1d9f404e71948311fad95978de : FAILURE in 2m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7eb84aad9210418c8b8b3e89cc3ba230 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec6c8a486a6d4cac9a6da7e659c07ef0 : SUCCESS in 3m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34429f3d9316b8025b125c2956a5086f1bce38f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-05-31 16:21:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"190dd3e8bb740ada48f79377b6f7b0127c4fd02b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 16:29:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/003b8e383da743e8b0437613f4383a70 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d260b460c9c34bba93a080366cb9a443 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7ab4debb6dc4b1f8f9ec55e0eac61f2 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd0780d222bc4195a42677358bab2052 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc3347fd0b4a4cac86e8459e7e3d9bc2 : SUCCESS in 3m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d56a8f0bb0b09865d5b5868291648b289f4ce4a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-05-31 17:00:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b6854da731565252dbd3d32648fd358343790a2a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-05-31 17:06:14.000000000","message":"Abandoned\n\nFixes have already submitted: https://review.opendev.org/c/openstack/keystone-specs/+/839945, https://review.opendev.org/c/openstack/keystone-specs/+/839909","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2e237c62798f264fa5d02b14aa178cb2f3cb2fd4","revisions":{"a5f58a35ce3c120b8278d7784a00d91e55a26423":{"kind":"REWORK","_number":1,"created":"2022-05-31 15:49:05.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/00/844100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/00/844100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/00/844100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/00/844100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/00/844100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/00/844100/1"}}},"commit":{"parents":[{"commit":"f9f4e50737a01409391517b9b5be049419223554","subject":"Describe the need for a default manager role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f9f4e50737a01409391517b9b5be049419223554"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-31 15:48:54.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-31 15:48:54.000000000","tz":540},"subject":"[WIP] Fix document build","message":"[WIP] Fix document build\n\nChange-Id: Idcfe2957fa83723f6de927ede36d21c7b5e6431e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/a5f58a35ce3c120b8278d7784a00d91e55a26423"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/a5f58a35ce3c120b8278d7784a00d91e55a26423"}]},"branch":"refs/heads/master"},"a0c8e326f0cbd9b1ff11ad7055fffa66b451cef3":{"kind":"REWORK","_number":2,"created":"2022-05-31 16:21:54.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/00/844100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/00/844100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/00/844100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/00/844100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/00/844100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/00/844100/2"}}},"commit":{"parents":[{"commit":"f9f4e50737a01409391517b9b5be049419223554","subject":"Describe the need for a default manager role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f9f4e50737a01409391517b9b5be049419223554"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-31 15:48:54.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-31 16:21:44.000000000","tz":540},"subject":"[WIP] Fix document build","message":"[WIP] Fix document build\n\nChange-Id: Idcfe2957fa83723f6de927ede36d21c7b5e6431e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/a0c8e326f0cbd9b1ff11ad7055fffa66b451cef3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/a0c8e326f0cbd9b1ff11ad7055fffa66b451cef3"}]},"branch":"refs/heads/master"},"2e237c62798f264fa5d02b14aa178cb2f3cb2fd4":{"kind":"REWORK","_number":3,"created":"2022-05-31 17:00:55.000000000","uploader":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"ref":"refs/changes/00/844100/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/00/844100/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/00/844100/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/00/844100/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/00/844100/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/00/844100/3"}}},"commit":{"parents":[{"commit":"f9f4e50737a01409391517b9b5be049419223554","subject":"Describe the need for a default manager role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f9f4e50737a01409391517b9b5be049419223554"}]}],"author":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-31 15:48:54.000000000","tz":540},"committer":{"name":"Hiromu Asahina","email":"hiromu.asahina.az@hco.ntt.co.jp","date":"2022-05-31 17:00:41.000000000","tz":540},"subject":"[WIP] Fix document build","message":"[WIP] Fix document build\n\nChange-Id: Idcfe2957fa83723f6de927ede36d21c7b5e6431e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/2e237c62798f264fa5d02b14aa178cb2f3cb2fd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/2e237c62798f264fa5d02b14aa178cb2f3cb2fd4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
