)]}'
{"id":"openstack%2Fkeystone-specs~851178","triplet_id":"openstack%2Fkeystone-specs~master~Iaba4a7f39fbc0ed26339cac4d5c693fa0684c7cb","project":"openstack/keystone-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"16465":{"account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"last_update":"2022-09-27 14:29:03.000000000","reason":"\u003cGERRIT_ACCOUNT_16465\u003e replied on the change","reason_account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},"35058":{"account":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"last_update":"2022-10-19 12:47:50.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2022-09-16 17:56:45.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2022-10-19 12:47:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaba4a7f39fbc0ed26339cac4d5c693fa0684c7cb","subject":"remove unicode from code","status":"MERGED","created":"2022-07-27 09:33:31.000000000","updated":"2022-10-19 12:48:53.000000000","submitted":"2022-10-19 12:47:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851178","meta_rev_id":"12dec1fcb3210d23243329d5f002341ba9267122","_number":851178,"virtual_id_number":851178,"owner":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-19 12:47:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-10-19 12:36:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2022-10-19 12:36:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2022-10-19 12:36:41.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2022-10-19 12:36:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2022-10-19 12:36:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-27 09:38:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-29 14:34:16.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-07-29 14:34:16.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2022-07-29 14:34:16.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2022-07-29 14:34:16.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2022-09-27 09:41:05.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"7705711ea460b5eabe42818a951fb424523f3e50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-07-27 09:33:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"228f962ea80b19879d3c36f5a3b376139217f1e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 09:38:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbf2585f49dc42fe97089b63926266df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e75fbabdf644b6781cf077c8aee4a25 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df6dec3a69184bb8907c8a978ad4bea5 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"83b67f2cec583d58f31b303eb7dfbad66f33d6e2","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-07-29 14:34:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e32d9eda081451352aeb331cbd923147bfc8ed10","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-20 11:13:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58b537d2d8f79918a6f0384dcedf1d1935adbb04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-20 11:18:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cf9bca0f2694278b1b919abc40caec8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed3768a4af884fdf9b686d4da05657e0 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78ae187d31fd4950a6177b173cc6d920 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"1ee065fa5a6d2299d37af8de293b37e028f8b1b9","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2022-09-16 17:56:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"64605bdd948ea853d92cafbc01106960e2b7648d","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-09-27 09:41:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cd1a88b9e54bc6722a5ec414747563a336bdff3a","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-09-27 14:29:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a1213523ff42c9808e12aa6fd4830356180ce9be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-10-19 12:36:41.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b883138dd017a240fd3a26e188d1b8ff95f1a07a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 12:41:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ec3e5252e564aa88c206ce72f30d36e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/957bbe3af5484f7f9621f25ce59dd237 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eeedff83d0e4033af34847fa41934aa : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"01f608e100b1541a7c02232736503db7d1f16f21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 12:42:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9de6ebd5377d930ce7d8960c13e71ef062ac6d75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 12:47:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b70686fa4eb248b7b46f0827566e61ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be80528795f54529920552debf83d1b1 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5436905b9494595a816b805a1256486 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":2},{"id":"985b883b892c13cca8dbe759c37127125eafd418","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 12:47:50.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":2},{"id":"12dec1fcb3210d23243329d5f002341ba9267122","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 12:48:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c73ae5016ad4fe4a499e238906fb9d9\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/8cadb06322304c16895bdb39f38164b8 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7071cf3e949cd7bee726b0794b3585cfe25a7f96","revisions":{"b52a31cf1deaeca3b12c259181e464476f65265a":{"kind":"REWORK","_number":1,"created":"2022-07-27 09:33:31.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/78/851178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/78/851178/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/78/851178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/78/851178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/78/851178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/78/851178/1"}}},"commit":{"parents":[{"commit":"491b4ee4ceba5238ad3f30fae793ba7586687be1","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/491b4ee4ceba5238ad3f30fae793ba7586687be1"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-27 09:32:45.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-27 09:32:58.000000000","tz":480},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: Iaba4a7f39fbc0ed26339cac4d5c693fa0684c7cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b52a31cf1deaeca3b12c259181e464476f65265a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/b52a31cf1deaeca3b12c259181e464476f65265a"}]},"branch":"refs/heads/master"},"7071cf3e949cd7bee726b0794b3585cfe25a7f96":{"kind":"NO_CHANGE","_number":2,"created":"2022-10-19 12:36:41.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/78/851178/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/78/851178/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/78/851178/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/78/851178/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/78/851178/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/78/851178/2"}}},"commit":{"parents":[{"commit":"24b290674aa84a4834130daa8057ebb0c71ba084","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/24b290674aa84a4834130daa8057ebb0c71ba084"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-27 09:32:45.000000000","tz":480},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2022-10-19 12:36:41.000000000","tz":0},"subject":"remove unicode from code","message":"remove unicode from code\n\nChange-Id: Iaba4a7f39fbc0ed26339cac4d5c693fa0684c7cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/7071cf3e949cd7bee726b0794b3585cfe25a7f96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/7071cf3e949cd7bee726b0794b3585cfe25a7f96"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[]}
