)]}'
{"id":"openstack%2Fkeystone-specs~938012","triplet_id":"openstack%2Fkeystone-specs~master~I23e71b07632f3d3ec705e2346c0d6c69c6643361","project":"openstack/keystone-specs","branch":"master","topic":"bug/2026345","attention_set":{},"removed_from_attention_set":{"37374":{"account":{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"},"last_update":"2025-01-10 14:22:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I23e71b07632f3d3ec705e2346c0d6c69c6643361","subject":"Replace `sphinxcontrib-*diag` with `graphviz`","status":"MERGED","created":"2024-12-18 18:44:17.000000000","updated":"2025-01-10 14:24:51.000000000","submitted":"2025-01-10 14:22:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938012-bug/2026345","meta_rev_id":"91f51f201cb1d4d24b7dd38797f17c38bb0ebe95","_number":938012,"virtual_id_number":938012,"owner":{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-10 14:22:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2025-01-08 15:32:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2025-01-10 14:12:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":1,"date":"2025-01-07 18:08:12.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2025-01-10 14:12:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-18 19:17:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-07 18:08:12.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-01-08 15:32:09.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2025-01-10 14:12:12.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"2719767fab12943af3a74cf7baec2bf95dc1ce5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"},"date":"2024-12-18 18:44:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d9ff0fd43f8b988a6c3cc1f3cfb8069e44b8565","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 19:17:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ae6c0acb46a4f009c8d61f8b8db2778\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28753d8e0d174c27bdba1a55e2a8fb1a : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3472dd777054468ab54ecd0a55b42e4d : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9fe49882d91469b991d85e5a03036fe : FAILURE in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1a805ddcdb645e38c0c268799c9b663 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f78325f523a2476d891ef0ccfb257059 : SUCCESS in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31002028f55374a2feaf00519984f9ee72e2b718","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"},"date":"2024-12-18 19:50:36.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"48b73ef37ee1e9b425098dc58823d9802039d1d4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"},"date":"2024-12-18 20:01:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8992de3bb69cfe4fc9865ba9ee3f234eb97732c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-18 20:06:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6972a75bf2c4d44bca2b857cab55a2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92fed1b33a8e460dbaa5ee345e2a9e83 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63eddf8f23794fa88b0dd15bdfe3d873 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6194a55672044f3837d00e177a2c5d8 : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/931fb5b746c94dd9a4c72b2a272919fe : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":2},{"id":"94cd120090579dbdbdc92941f1b6a488f2d2ceca","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"},"date":"2024-12-18 20:09:51.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"3545886322ef81ee09d0d02fa6fc334f87daaee5","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-01-07 18:08:12.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a26a6eb00aa10c5b0eb2b3d11d75071496f85e90","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-01-08 15:32:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"23f432862f6d8535ecb75a5f0ae37e96c7f947c7","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-01-10 14:12:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2bda7df1c51cfe9086cac226f12936d54135f1f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 14:12:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5166ac8ae953be9a4e8dc038d3673fd7def2ef69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 14:22:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d93a6e815607489380d7e8c8edb3e173\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73219bf1beb54c18a8160c4267e00d49 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1c742dba27e4d4c8fbf7bd1f2e0b423 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db62cca1efcc4f5ab95b2cbc6a131ea2 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/748f6d9797b24106a09d5eeca342dc8b : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":2},{"id":"61dd97bff3eb91b7a41e998f82e7910207e2db14","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 14:22:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"91f51f201cb1d4d24b7dd38797f17c38bb0ebe95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 14:24:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18021d6ebe4c48a1811f336b11ddd8cd\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/c9200f52dd234b578fdbe150fa00812d : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"46c7b05459de06327053fe31eb3eb569936383dd","revisions":{"f48f4958a9fcb26dd902a7ceac9856df78d75bad":{"kind":"REWORK","_number":1,"created":"2024-12-18 18:44:17.000000000","uploader":{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"},"ref":"refs/changes/12/938012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/12/938012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/12/938012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/12/938012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/12/938012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/12/938012/1"}}},"commit":{"parents":[{"commit":"fc91ff557c756704842f21c6572ce5c802354443","subject":"Merge \"Add identity spec for Domain Manager persona\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fc91ff557c756704842f21c6572ce5c802354443"}]}],"author":{"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","date":"2024-12-18 18:14:09.000000000","tz":60},"committer":{"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","date":"2024-12-18 18:43:21.000000000","tz":60},"subject":"Replace `sphinxcontrib-*diag` with `graphviz`","message":"Replace `sphinxcontrib-*diag` with `graphviz`\n\nReplace unmaintained `sphinxcontrib-*diag` with `graphviz` in order to\nfix not building docs (e.g. with `tox -e docs` or in CI) after upgrade\nof `pillow` in upper-constraints.txt of openstack/requirements [1]\n(pulled as dependency of `sphinxcontrib-*diag` [2]).\nThe change is similar to how other OpenStack services did it (e.g.\nKeystone [3] + Partial-Bug #2026345).\n\nSome diagrams (specifically `seqdiag`) is difficult to migrate\npreserving readability, thus they are replaced with corresponding static\nimages from the latest published build `0.0.1.dev621`, given that specs\nare not expected to be updated once merged.\n\n[1] https://opendev.org/openstack/requirements/commit/64747a5e55d48caa5ce2fb3cfbc76e3665293b32\n[2] https://github.com/blockdiag/sphinxcontrib-blockdiag/issues/26\n[3] https://review.opendev.org/c/openstack/keystone/+/935685\n\nPartial-Bug: #2026345\nChange-Id: I23e71b07632f3d3ec705e2346c0d6c69c6643361\nSigned-off-by: Stanislav Zaprudskiy \u003cs.zaprudskiy@sap.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f48f4958a9fcb26dd902a7ceac9856df78d75bad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f48f4958a9fcb26dd902a7ceac9856df78d75bad"}]},"branch":"refs/heads/master"},"46c7b05459de06327053fe31eb3eb569936383dd":{"kind":"REWORK","_number":2,"created":"2024-12-18 20:01:52.000000000","uploader":{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"},"ref":"refs/changes/12/938012/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/12/938012/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/12/938012/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/12/938012/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/12/938012/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/12/938012/2"}}},"commit":{"parents":[{"commit":"fc91ff557c756704842f21c6572ce5c802354443","subject":"Merge \"Add identity spec for Domain Manager persona\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/fc91ff557c756704842f21c6572ce5c802354443"}]}],"author":{"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","date":"2024-12-18 18:14:09.000000000","tz":60},"committer":{"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","date":"2024-12-18 20:01:28.000000000","tz":60},"subject":"Replace `sphinxcontrib-*diag` with `graphviz`","message":"Replace `sphinxcontrib-*diag` with `graphviz`\n\nReplace unmaintained `sphinxcontrib-*diag` with `graphviz` in order to\nfix not building docs (e.g. with `tox -e docs` or in CI) after upgrade\nof `pillow` in upper-constraints.txt of openstack/requirements [1]\n(pulled as dependency of `sphinxcontrib-*diag` [2]).\nThe change is similar to how other OpenStack services did it (e.g.\nKeystone [3] + Partial-Bug #2026345).\n\nSome diagrams (specifically `seqdiag`) is difficult to migrate\npreserving readability, thus they are replaced with corresponding static\nimages from the latest published build `0.0.1.dev621`, given that specs\nare not expected to be updated once merged.\n\nOther included changes are to fix the gates, and also require some\nrecent supported Python versions.\n\n[1] https://opendev.org/openstack/requirements/commit/64747a5e55d48caa5ce2fb3cfbc76e3665293b32\n[2] https://github.com/blockdiag/sphinxcontrib-blockdiag/issues/26\n[3] https://review.opendev.org/c/openstack/keystone/+/935685\n\nPartial-Bug: #2026345\nChange-Id: I23e71b07632f3d3ec705e2346c0d6c69c6643361\nSigned-off-by: Stanislav Zaprudskiy \u003cs.zaprudskiy@sap.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/46c7b05459de06327053fe31eb3eb569936383dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/46c7b05459de06327053fe31eb3eb569936383dd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
