)]}'
{"id":"openstack%2Fkeystone-specs~974616","triplet_id":"openstack%2Fkeystone-specs~master~Ida2fcd8ae91eaa6f0d5f90c6cf5f23d44e49b3f7","project":"openstack/keystone-specs","branch":"master","hashtags":[],"change_id":"Ida2fcd8ae91eaa6f0d5f90c6cf5f23d44e49b3f7","subject":"tox: Drop basepython","status":"NEW","created":"2026-01-24 14:06:23.000000000","updated":"2026-01-24 14:17:50.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"682e5286417031436da82d17fbed75570a0b274f","_number":974616,"virtual_id_number":974616,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-01-24 14:17:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-24 14:17:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b3d8dd8374fe471ef21b12536bec10ffd27dd76","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 14:06:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7934f42838a184dd699ee5afc70aa3300a3f0dae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 14:06:58.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ace3db3f45747bfa009e4f4348e1fa98d900acaa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 14:07:14.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"fd1c57c823b6735548d05f81a1ea8a5e6391d635","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 14:07:17.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"682e5286417031436da82d17fbed75570a0b274f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 14:17:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6b05f57c685426bb07cf8ea9595ab30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f880bc3a6969482db5b216ad1c2cc31a : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3436afbf9d864f03896798740b89f185 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86326dbbf2fb4ac7ae1604438a384dd1 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/69ce8c58b2e44cfa9e0b2fb950c92aef : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"9cdbc52c8b3ffd1d926835493ef531d246bfd591","revisions":{"9ee71428dce8f081dc98b7e8ca4d378db1d1c07a":{"kind":"REWORK","_number":1,"created":"2026-01-24 14:06:23.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/16/974616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/16/974616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/974616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/974616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/974616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/16/974616/1"}}},"commit":{"parents":[{"commit":"2cc6c7c8a59a303098aaccaa776352c13b79d1d3","subject":"Claim `Include invalid password details in audit messages`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/2cc6c7c8a59a303098aaccaa776352c13b79d1d3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 14:06:23.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 14:06:23.000000000","tz":0},"subject":"tox: Drop basepython","message":"tox: Drop basepython\n\nPython 2 reached its EOL long time ago and we no longer expect any\nuser may attempt to run tox in Python 2.\n\nRemoving the option allows us to remove ignore_basepython_conflict.\n\nChange-Id: Ida2fcd8ae91eaa6f0d5f90c6cf5f23d44e49b3f7\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/9ee71428dce8f081dc98b7e8ca4d378db1d1c07a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/9ee71428dce8f081dc98b7e8ca4d378db1d1c07a"}]},"branch":"refs/heads/master"},"1bc1dc22f6d68dce2159d5839a6ddc69b5510123":{"kind":"REWORK","_number":2,"created":"2026-01-24 14:06:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/16/974616/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/16/974616/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/974616/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/974616/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/974616/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/16/974616/2"}}},"commit":{"parents":[{"commit":"2cc6c7c8a59a303098aaccaa776352c13b79d1d3","subject":"Claim `Include invalid password details in audit messages`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/2cc6c7c8a59a303098aaccaa776352c13b79d1d3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 14:06:23.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 14:06:58.000000000","tz":0},"subject":"tox: Drop basepython","message":"tox: Drop basepython\n\nPython 2 reached its EOL long time ago and we no longer expect any\nuser may attempt to run tox in Python 2.\n\nRemoving the option allows us to remove ignore_basepython_conflict.\n\nChange-Id: Ida2fcd8ae91eaa6f0d5f90c6cf5f23d44e49b3f7\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/1bc1dc22f6d68dce2159d5839a6ddc69b5510123"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/1bc1dc22f6d68dce2159d5839a6ddc69b5510123"}]},"branch":"refs/heads/master"},"9cdbc52c8b3ffd1d926835493ef531d246bfd591":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-24 14:07:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/16/974616/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/16/974616/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/974616/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/974616/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/16/974616/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/16/974616/3"}}},"commit":{"parents":[{"commit":"2cc6c7c8a59a303098aaccaa776352c13b79d1d3","subject":"Claim `Include invalid password details in audit messages`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/2cc6c7c8a59a303098aaccaa776352c13b79d1d3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 14:06:23.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 14:07:14.000000000","tz":0},"subject":"tox: Drop basepython","message":"tox: Drop basepython\n\nPython 2 reached its EOL long time ago and we no longer expect any\nuser may attempt to run tox in Python 2.\n\nRemoving the option allows us to remove ignore_basepython_conflict and bump minversion.\n\nChange-Id: Ida2fcd8ae91eaa6f0d5f90c6cf5f23d44e49b3f7\nCo-authored-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/9cdbc52c8b3ffd1d926835493ef531d246bfd591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/9cdbc52c8b3ffd1d926835493ef531d246bfd591"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
