)]}'
{"id":"openstack%2Fkeystone-specs~975773","triplet_id":"openstack%2Fkeystone-specs~master~I4154ffb0a7b5e473acc92861ef2279b83fd83f92","project":"openstack/keystone-specs","branch":"master","attention_set":{"13478":{"account":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"last_update":"2026-02-05 11:07:33.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I4154ffb0a7b5e473acc92861ef2279b83fd83f92","subject":"application credentials compliance options","status":"NEW","created":"2026-02-05 10:57:58.000000000","updated":"2026-04-21 11:14:28.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3414141d03b057ecac1224fd7181de655fa05dc5","_number":975773,"virtual_id_number":975773,"owner":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":26892,"name":"Marcus Murwall","email":"marcus.murwall@cleura.com","username":"muran"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-05 13:03:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2026-04-21 11:04:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":1,"date":"2026-04-21 11:14:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":26892,"name":"Marcus Murwall","email":"marcus.murwall@cleura.com","username":"muran"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":26892,"name":"Marcus Murwall","email":"marcus.murwall@cleura.com","username":"muran"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26892,"name":"Marcus Murwall","email":"marcus.murwall@cleura.com","username":"muran"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-05 11:07:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-21 11:04:19.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2026-04-21 11:14:28.000000000","updated_by":{"_account_id":26892,"name":"Marcus Murwall","email":"marcus.murwall@cleura.com","username":"muran"},"reviewer":{"_account_id":26892,"name":"Marcus Murwall","email":"marcus.murwall@cleura.com","username":"muran"},"state":"REVIEWER"}],"messages":[{"id":"c3c2d21cfdf8e537b1ba80c050edd366d1d383fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2026-02-05 10:57:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b98a8472fa1d3ea9824c0271e4626561813de67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 11:07:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c205c19177044af98ea9c1098847cf63\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df988dc99e04bd8bcd6cd62e5268877 : FAILURE in 3m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e84caf0bbbfa48349cb4069dc75aab5d : FAILURE in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"10c23b35d815f303cf79b1d6c8a4deb31f6fd350","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2026-02-05 12:54:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6786f8cd770b992f31889c53f4626442944bb4a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 13:03:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74370313221a42b889486fbeae1e1a92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/553e5e9a22c6421680a1088270861be7 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cb7a7cf0cf445f8a86d01e5366bc219 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"8fa4a9d6e8a08b22d8f3590b64e2467ee0686fca","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-21 11:04:19.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3414141d03b057ecac1224fd7181de655fa05dc5","author":{"_account_id":26892,"name":"Marcus Murwall","email":"marcus.murwall@cleura.com","username":"muran"},"date":"2026-04-21 11:14:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0fe9e16eea317f2a5bd6b1ae612c8de0d7cd8948","revisions":{"f8f1cd3882a9ebead314e58709d4a39d69dfb6ec":{"kind":"REWORK","_number":1,"created":"2026-02-05 10:57:58.000000000","uploader":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"ref":"refs/changes/73/975773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/73/975773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/73/975773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/73/975773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/73/975773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/73/975773/1"}}},"commit":{"parents":[{"commit":"2cc6c7c8a59a303098aaccaa776352c13b79d1d3","subject":"Claim `Include invalid password details in audit messages`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/2cc6c7c8a59a303098aaccaa776352c13b79d1d3"}]}],"author":{"name":"Boris Bobrov","email":"b.bobrov@sap.com","date":"2026-02-05 10:52:16.000000000","tz":60},"committer":{"name":"Boris Bobrov","email":"b.bobrov@sap.com","date":"2026-02-05 10:57:17.000000000","tz":60},"subject":"application credentials compliance options","message":"application credentials compliance options\n\nAdd two new configuration options in [application_credential] section:\n- disallow_user_secrets: Prevent users from providing their own secret\n- max_lifetime_days: Enforce maximum lifetime for application credentials\n\nThese options help cloud operators enforce compliance requirements for\napplication credentials.\n\nPartial-Bug: 2140492\nGenerated-By: Claude (Anthropic)\nChange-Id: I4154ffb0a7b5e473acc92861ef2279b83fd83f92\nSigned-off-by: Boris Bobrov \u003cb.bobrov@sap.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f8f1cd3882a9ebead314e58709d4a39d69dfb6ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/f8f1cd3882a9ebead314e58709d4a39d69dfb6ec"}]},"branch":"refs/heads/master"},"0fe9e16eea317f2a5bd6b1ae612c8de0d7cd8948":{"kind":"REWORK","_number":2,"created":"2026-02-05 12:54:54.000000000","uploader":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"ref":"refs/changes/73/975773/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-specs","ref":"refs/changes/73/975773/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/73/975773/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/73/975773/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-specs refs/changes/73/975773/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-specs refs/changes/73/975773/2"}}},"commit":{"parents":[{"commit":"2cc6c7c8a59a303098aaccaa776352c13b79d1d3","subject":"Claim `Include invalid password details in audit messages`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/2cc6c7c8a59a303098aaccaa776352c13b79d1d3"}]}],"author":{"name":"Boris Bobrov","email":"b.bobrov@sap.com","date":"2026-02-05 10:52:16.000000000","tz":60},"committer":{"name":"Boris Bobrov","email":"b.bobrov@sap.com","date":"2026-02-05 12:54:41.000000000","tz":60},"subject":"application credentials compliance options","message":"application credentials compliance options\n\nAdd two new configuration options in [application_credential] section:\n- disallow_user_secrets: Prevent users from providing their own secret\n- max_lifetime_days: Enforce maximum lifetime for application credentials\n\nThese options help cloud operators enforce compliance requirements for\napplication credentials.\n\nPartial-Bug: 2140492\nGenerated-By: Claude (Anthropic)\nChange-Id: I4154ffb0a7b5e473acc92861ef2279b83fd83f92\nSigned-off-by: Boris Bobrov \u003cb.bobrov@sap.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/0fe9e16eea317f2a5bd6b1ae612c8de0d7cd8948"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-specs/commit/0fe9e16eea317f2a5bd6b1ae612c8de0d7cd8948"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
