)]}'
{"id":"openstack%2Fkeystone-tempest-plugin~966128","triplet_id":"openstack%2Fkeystone-tempest-plugin~master~I9586938ab1e9012f72bd3d2240c9aacae04499a1","project":"openstack/keystone-tempest-plugin","branch":"master","topic":"drop-py39","attention_set":{},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2025-12-15 13:44:28.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2025-12-15 13:09:20.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2025-12-15 13:44:28.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-11-25 19:33:45.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"I9586938ab1e9012f72bd3d2240c9aacae04499a1","subject":"Drop python 3.9 and support python 3.13","status":"MERGED","created":"2025-11-04 21:01:04.000000000","updated":"2025-12-15 13:44:28.000000000","submitted":"2025-12-15 13:44:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966128-drop-py39","meta_rev_id":"3ab4183a409a7231532bd107247383a101647557","_number":966128,"virtual_id_number":966128,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-15 13:44:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":2,"date":"2025-12-15 13:09:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":2,"date":"2025-11-25 19:33:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-11 08:43:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":1,"date":"2025-12-15 13:09:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-04 21:48:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-25 19:31:18.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-12-11 08:43:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-12-11 08:43:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2025-12-11 08:43:12.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"b5725ff218f0d4aefc68cb00ff2798b97fca1569","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2025-11-04 21:01:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"32c8300086bbbbf29da96d6fb9f68b8d0153ccc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-04 21:48:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2b85294ff944fe389cd03f95a325cb3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/044018f06d4142ee9a3b6b485a219a52 : SUCCESS in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8395b81cebc4e27beeb34a0072f1c70 : SUCCESS in 2m 09s\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/d33924568ae1456d9c0954826ddc6734 : SUCCESS in 26m 55s\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/3f4f05307ce646d19817956bf5c496ba : FAILURE in 30m 57s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/5a2f618130324617ae7f5700cd5916e5 : SUCCESS in 18m 43s\n- keystone-tempest-2025-2 https://zuul.opendev.org/t/openstack/build/ea5b6c228bc94fceb3a9b76375b4bcda : SUCCESS in 31m 47s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/d1e3e97834934ca8bd2d55e9b4b11857 : FAILURE in 42m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7cf0d5cc7d05fef2b634ec101e65937355d8b23c","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-11-25 19:33:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bb6f7d120c1ea93c46c41b2c07b004da24045e8d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-11 08:43:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f92288d324410d76dc5b9148ae46efa93b1767c3","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-12-15 13:09:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bea164d0e2cf59095931d22dc8127dd213b0bbe1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 13:09:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32c33fabfdc5198e8fb00c55f0045a959df0730d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 13:44:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57a3bee24a944bcdb57a48fd79c0fc0b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7200524b15334d0d810694451d9d4a31 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/303f80108b6c4d27b3c4ddead273682e : SUCCESS in 2m 14s\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/edf1ba686fb540088096ce9df11baf44 : SUCCESS in 29m 41s\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/fc9ea6ba0b9d4a45ac005d94939cec1c : SUCCESS in 13m 48s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/a445d0b905224a638f7d09ccb5f2c121 : FAILURE in 20m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ab4183a409a7231532bd107247383a101647557","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-15 13:44:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a6921676c98c4cb0c8965bffecfaf54a7ef9c8b9","revisions":{"a6921676c98c4cb0c8965bffecfaf54a7ef9c8b9":{"kind":"REWORK","_number":1,"created":"2025-11-04 21:01:04.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/28/966128/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone-tempest-plugin","ref":"refs/changes/28/966128/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone-tempest-plugin refs/changes/28/966128/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone-tempest-plugin refs/changes/28/966128/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone-tempest-plugin refs/changes/28/966128/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone-tempest-plugin refs/changes/28/966128/1"}}},"commit":{"parents":[{"commit":"97d6de342fc0b7487c2dc99765c4717850758afc","subject":"Add stable/2025.2 job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-tempest-plugin/commit/97d6de342fc0b7487c2dc99765c4717850758afc"}]}],"author":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-04 20:53:28.000000000","tz":0},"committer":{"name":"Ghanshyam Maan","email":"gmaan@ghanshyammann.com","date":"2025-11-04 21:01:03.000000000","tz":0},"subject":"Drop python 3.9 and support python 3.13","message":"Drop python 3.9 and support python 3.13\n\nDue to requirements repo dropped the constraints for\npython3.9[1], Tempest needs to drop the python 3.9\nsupport[2]. Because Tempest master cannot support\npy3.9 anymore, it needs to be dropped from tempest\nplugins also.\n\nAlso, add python 3.13 support which is tested as mandatory\nin 2026.2 cycle.\n\n[1] https://review.opendev.org/c/openstack/requirements/+/957587\n[2] https://review.opendev.org/c/openstack/tempest/+/966101\n\nChange-Id: I9586938ab1e9012f72bd3d2240c9aacae04499a1\nSigned-off-by: Ghanshyam Maan \u003cgmaan@ghanshyammann.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-tempest-plugin/commit/a6921676c98c4cb0c8965bffecfaf54a7ef9c8b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone-tempest-plugin/commit/a6921676c98c4cb0c8965bffecfaf54a7ef9c8b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
