)]}'
{"id":"openstack%2Fkeystoneauth~1003778","triplet_id":"openstack%2Fkeystoneauth~master~I10443e5c79c49af7fc51888f77ed434636fe1d2b","project":"openstack/keystoneauth","branch":"master","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-03 15:31:40.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I10443e5c79c49af7fc51888f77ed434636fe1d2b","subject":"trivial: Remove type info from docstrings","status":"NEW","created":"2026-09-03 13:48:17.000000000","updated":"2026-09-03 15:31:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a0db268045ce78d0afaf1b0f0bf8d35e5015a043","_number":1003778,"virtual_id_number":1003778,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-09-03 15:31:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 15:31:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8cea90d6dbae67ab4b8a4a2ba3e0a3340af3991","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 13:48:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0db268045ce78d0afaf1b0f0bf8d35e5015a043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:31:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2f128ffc4e843d0ad5454f3836c93d4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a506d2e7774c4585bda84417a4937c39 : SUCCESS in 45m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69a0f43396e243388ba132ec9ddf4494 : SUCCESS in 8m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d7ca19553154899aabc24d9a39ab831 : SUCCESS in 6m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f7e70fab8c44e48822f2ba531771aa4 : SUCCESS in 4m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/e4228058f19b40d39fe5589b9e622ed2 : SUCCESS in 11m 21s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/5f7acea8bbd64b6c9049a3a1894b836c : SUCCESS in 12m 54s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/62ed84475c184eab90b76882783746e5 : SUCCESS in 1h 17m 23s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/966de75b7906472ea1d7f5382e52a92e : SUCCESS in 4m 35s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/da22340633734f37890d1154642f2dad : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c4960b50ccf4c83a7bcc10651b2fd21 : FAILURE in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c2eb6a6b6b547e095fcd9e5ef9ddf8c : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0192de40838478ce3819292241839825cab75e8b","revisions":{"0192de40838478ce3819292241839825cab75e8b":{"kind":"REWORK","_number":1,"created":"2026-09-03 13:48:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/78/1003778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/78/1003778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/78/1003778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/78/1003778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/78/1003778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/78/1003778/1"}}},"commit":{"parents":[{"commit":"eab194aadb5ee2d930650895daee4ddd8b3ce418","subject":"typing: Remove unnecessary use of TYPE_CHECKING","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/eab194aadb5ee2d930650895daee4ddd8b3ce418"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-03 09:18:55.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-03 09:18:55.000000000","tz":60},"subject":"trivial: Remove type info from docstrings","message":"trivial: Remove type info from docstrings\n\nThis is unnecessary since the codebase is typed.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I10443e5c79c49af7fc51888f77ed434636fe1d2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/0192de40838478ce3819292241839825cab75e8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/0192de40838478ce3819292241839825cab75e8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
