)]}'
{"id":"openstack%2Fkeystoneauth~1003780","triplet_id":"openstack%2Fkeystoneauth~master~I7df3c4bf3101aaa29b8081ae519dd2d9037717df","project":"openstack/keystoneauth","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-03 15:17:54.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I7df3c4bf3101aaa29b8081ae519dd2d9037717df","subject":"typing: Allow None passed to set_auth_state","status":"MERGED","created":"2026-09-03 13:49:43.000000000","updated":"2026-09-03 16:54:05.000000000","submitted":"2026-09-03 16:52:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1003780","meta_rev_id":"ab4d10e8a4f8d349bc3cf9906e89ff741b2b640c","_number":1003780,"virtual_id_number":1003780,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-03 16:52:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-09-03 14:59:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2026-09-03 15:17:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2026-09-03 15:17:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-03 14:59:09.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2026-09-03 15:15:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b1750f4a171095a002a4afc17219270937884c37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 13:49:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7933053575bb6d33941545724ad0d5937d8795e2","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-03 14:59:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0a6ce51d45a6c6f7ab10bf750d252c2eb7568192","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:15:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/417bd3a5353c4cf48fa7fff45ab5dfa3\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb5eb4b76910476d833356b677341bc8 : SUCCESS in 1h 05m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/622647b950634738a6d7759c8428ea2a : SUCCESS in 8m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3cd06cfab0c44701ba8ddffce2feb21c : SUCCESS in 7m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/09200a9bc1e24e69af166ab9c81d71d4 : SUCCESS in 8m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1a3c4452751f43b9bbacf1ed4d4ee029 : SUCCESS in 10m 33s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/2da125de1801414b88fc1af04c4bf7f1 : SUCCESS in 16m 27s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a5609f685dd942259022492a4957f13b : SUCCESS in 50m 42s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f0c6cdaa0cd24742a48cf8de575bab04 : SUCCESS in 5m 42s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/98b5cd87144c46968292a0022a4547a5 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/312acf1cfeea4896b9d94c01377d0f70 : SUCCESS in 12m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3be6fbaeb106472d87faf82262962f8e : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"339bb35adbf5059234a45bbd7c29e9417a213045","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 15:17:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e5e0cb6e0e89298672d9c813744ea09a7c807a18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:18:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b151644f6c5c58c54ef4a9e09fcb782bcb41542","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 16:52:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cec47fd187d74b8e8e50462cef26e3bb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/67add761559b4ea7bd7d4e8f89aac8ab : SUCCESS in 1h 32m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8edf09a78b0f4e97b34406c56cb6429b : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/60d3a7833661473da19300e8a4fde43f : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b38875ef071e4aa4a4493e51d5c4d534 : SUCCESS in 9m 26s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9634f86e0e4c4370b54b15ee9635752c : SUCCESS in 6m 44s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f8223f167afd4acc8857a5be99156758 : SUCCESS in 35m 50s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f6eb7b4a13704ab3bc1c473a2ef17129 : SUCCESS in 3m 40s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/f7b077ea919a421da35f274f79319d05 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2cf3bb859904bbda0749b3c5195091c : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8027c2abd79a4c969f1ff65aece1ebbf : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"140aea138faa85283af439e57d41d6c39824bf8c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 16:52:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ab4d10e8a4f8d349bc3cf9906e89ff741b2b640c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 16:54:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/084369ff8cef42d1b9d5b960a2f733f1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4644d608bc949acbf547ad7d575be98 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e91b3e3d5548e4193e86f9500eecb592b28c5831","revisions":{"e91b3e3d5548e4193e86f9500eecb592b28c5831":{"kind":"REWORK","_number":1,"created":"2026-09-03 13:49:43.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/80/1003780/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/80/1003780/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/80/1003780/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/80/1003780/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/80/1003780/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/80/1003780/1"}}},"commit":{"parents":[{"commit":"2ad12864d28b460f2193d891584e2873cb9724fa","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/2ad12864d28b460f2193d891584e2873cb9724fa"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-03 13:49:30.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-03 13:49:30.000000000","tz":60},"subject":"typing: Allow None passed to set_auth_state","message":"typing: Allow None passed to set_auth_state\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I7df3c4bf3101aaa29b8081ae519dd2d9037717df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/e91b3e3d5548e4193e86f9500eecb592b28c5831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/e91b3e3d5548e4193e86f9500eecb592b28c5831"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
