)]}'
{"id":"openstack%2Fkeystoneauth~1004368","triplet_id":"openstack%2Fkeystoneauth~master~Ic239a1402576f630d9d3d23c9129b334870365cc","project":"openstack/keystoneauth","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-07 10:02:09.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-09-07 11:35:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic239a1402576f630d9d3d23c9129b334870365cc","subject":"Fix typo","status":"MERGED","created":"2026-09-07 00:12:31.000000000","updated":"2026-09-07 11:36:32.000000000","submitted":"2026-09-07 11:35:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1004368","meta_rev_id":"a1e1921a09a7a8fd5e52bd4b53b9b49f191eb96f","_number":1004368,"virtual_id_number":1004368,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-07 11:35:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-07 10:02:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-07 10:02:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-07 00:13:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-07 01:31:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7d125aa3fd6d7c1997a0d54925e22d0ca984ed7d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:12:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f072667eedd8e83ea0a875ba2f6dd6f2bf91f9fc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:13:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3870e5f0c8040d719dcddcb9a7225236e789321c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:13:33.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"fe9a2fe111bfc4def9d3787ddf20f99c00285468","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:26:08.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"0b51e4a4deb7846ecd5e912ee8b6b86050f724f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-09-07 00:26:40.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"95dee83bb38de9feca60a196b3b0c8ada3c0ecc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 01:31:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/665ab9ba6afd4e49a9c53b8aa7c01d26\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d83b5d7ad6a841a29e4a1fbae274ccf8 : SUCCESS in 47m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19a3ef26be9e4cf785ec441d00233b36 : SUCCESS in 7m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/99f2ee06139f4661a47e36edcc3ddc9b : SUCCESS in 7m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b9f9f6fdf6e640089fdcd4823b448c82 : SUCCESS in 7m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3f8f985138c5445f80974873efded284 : SUCCESS in 9m 55s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/49b749393b9748c7bd9d6a48b9032740 : SUCCESS in 14m 40s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/cc639df2fc63413aa37185973ad73311 : SUCCESS in 59m 14s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/bae30693d4434d72b11da1ce5c8deedc : SUCCESS in 4m 58s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/6bd0a3db14904e6c9154befe622c7b4a : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54e1f87ba28f4100bf7b32062772314b : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6e5952cf55c4e7e859c1f91be2e75f6 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":4},{"id":"c3a5365e8b9773301515d6f8de18640e476dda77","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-07 10:02:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"30ad1f17b1d451324b5d7db34a6bed2756d93fd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 10:02:59.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"bf8ea6ce0ae5fb4be11e3fb1cc51b7ef9f48e6cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 11:35:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/835d57844d314ff2a324d6c0eed7f658\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c221b9d6d24470da605884ea40a65aa : SUCCESS in 1h 29m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dfb7fb3bfcd4491a62bc36c11514a92 : SUCCESS in 7m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd49bdf1e3d645d491530c270d20b9ea : SUCCESS in 7m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/35cd79fefcdd43ea96c31041e7d7b12a : SUCCESS in 7m 44s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7f08c54e7da24bd789383dac51dc8ccf : SUCCESS in 9m 29s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1d8aff29fdfb400a9c180027ee36f038 : SUCCESS in 1h 00m 38s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/abfc9e16add94d5ba2f354e7da37725a : SUCCESS in 6m 28s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/97b19d573ecf4ec2b8cc611fb68a3af0 : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33bf650f8fa749e1a40876f9903ef3ec : SUCCESS in 8m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be3b4d056e1d41c8845581f9ec2792f4 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":4},{"id":"db3ca43cb2c43e61c8b4d60d0c73b849e39406c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 11:35:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"a1e1921a09a7a8fd5e52bd4b53b9b49f191eb96f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-07 11:36:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b06d0aa730fa49de96c8328844686d22\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eff57798107466b8b124ac64e596f3e : SUCCESS in 56s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7524c0e2fb9997799cce8d3b7862b6ef2f7b08f1","revisions":{"512661dddf17779cf29d7c7c4490ace7ce740fb9":{"kind":"REWORK","_number":1,"created":"2026-09-07 00:12:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/68/1004368/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/68/1004368/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/68/1004368/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/68/1004368/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/68/1004368/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/68/1004368/1"}}},"commit":{"parents":[{"commit":"4741636c9075a7eefe5f13c2fa0796b5fd8508a5","subject":"feat: add Session.close() for deterministic resource release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/4741636c9075a7eefe5f13c2fa0796b5fd8508a5"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:12:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:12:31.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Ic239a1402576f630d9d3d23c9129b334870365cc\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/512661dddf17779cf29d7c7c4490ace7ce740fb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/512661dddf17779cf29d7c7c4490ace7ce740fb9"}]},"branch":"refs/heads/master","conflicts":{"no_base_reason":"NO_MERGE_PERFORMED","contains_conflicts":false}},"138bdb4da6984b1d5e4f8962e8b0827427e145f1":{"kind":"REWORK","_number":2,"created":"2026-09-07 00:13:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/68/1004368/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/68/1004368/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/68/1004368/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/68/1004368/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/68/1004368/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/68/1004368/2"}}},"commit":{"parents":[{"commit":"4741636c9075a7eefe5f13c2fa0796b5fd8508a5","subject":"feat: add Session.close() for deterministic resource release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/4741636c9075a7eefe5f13c2fa0796b5fd8508a5"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:12:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:13:24.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Ic239a1402576f630d9d3d23c9129b334870365cc\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/138bdb4da6984b1d5e4f8962e8b0827427e145f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/138bdb4da6984b1d5e4f8962e8b0827427e145f1"}]},"branch":"refs/heads/master"},"451e57f86dd2ba8b2823fb235b1addb2eb3a4aea":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-09-07 00:26:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/68/1004368/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/68/1004368/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/68/1004368/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/68/1004368/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/68/1004368/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/68/1004368/3"}}},"commit":{"parents":[{"commit":"4741636c9075a7eefe5f13c2fa0796b5fd8508a5","subject":"feat: add Session.close() for deterministic resource release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/4741636c9075a7eefe5f13c2fa0796b5fd8508a5"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:12:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:26:08.000000000","tz":0},"subject":"Fix typos","message":"Fix typos\n\nChange-Id: Ic239a1402576f630d9d3d23c9129b334870365cc\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/451e57f86dd2ba8b2823fb235b1addb2eb3a4aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/451e57f86dd2ba8b2823fb235b1addb2eb3a4aea"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7524c0e2fb9997799cce8d3b7862b6ef2f7b08f1":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-09-07 00:26:40.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/68/1004368/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/68/1004368/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/68/1004368/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/68/1004368/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/68/1004368/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/68/1004368/4"}}},"commit":{"parents":[{"commit":"4741636c9075a7eefe5f13c2fa0796b5fd8508a5","subject":"feat: add Session.close() for deterministic resource release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/4741636c9075a7eefe5f13c2fa0796b5fd8508a5"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:12:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-09-07 00:26:40.000000000","tz":0},"subject":"Fix typo","message":"Fix typo\n\nChange-Id: Ic239a1402576f630d9d3d23c9129b334870365cc\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/7524c0e2fb9997799cce8d3b7862b6ef2f7b08f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/7524c0e2fb9997799cce8d3b7862b6ef2f7b08f1"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
