)]}'
{"id":"openstack%2Fkeystoneauth~1005688","triplet_id":"openstack%2Fkeystoneauth~master~Ied2aa68bb461c2166053ef9e38a0cbe856316a26","project":"openstack/keystoneauth","branch":"master","topic":"fix/oidc-device-interval","attention_set":{},"removed_from_attention_set":{"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2026-09-21 16:49:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ied2aa68bb461c2166053ef9e38a0cbe856316a26","subject":"Default OIDC device authorization polling interval to 5s","status":"MERGED","created":"2026-09-15 09:26:34.000000000","updated":"2026-09-21 16:49:16.000000000","submitted":"2026-09-21 16:49:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"1005688-fix/oidc-device-interval","meta_rev_id":"4fb63aa346ea6cdb9462f73113cec752abc524d0","_number":1005688,"virtual_id_number":1005688,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-21 16:49:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-09-21 14:22:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2026-09-15 09:37:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-09-21 14:22:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 09:37:34.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2026-09-15 10:57:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-21 14:13:14.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"44fe43d361d6de0448bb759446127205c18fab81","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2026-09-15 09:26:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa5466ae8669b79030d00a46131618e269f51268","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2026-09-15 09:27:09.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"037600b097e17e74756022eba1f3cb18486fff83","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 09:37:34.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e2ac5e358d3c604d8144395d5a4ed821015a22c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 10:57:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25d732835a934c5998534c34ec4e3a8f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/997cee5c72cb4915a859fc3396424a50 : SUCCESS in 1h 21m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f08bde8251f84bb79add345caee2f498 : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64f9c121ccd146dba9a9e45bf0db19aa : SUCCESS in 8m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1f20b883edbc45098353223c2a56fd63 : SUCCESS in 4m 48s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d2c516ccf88f41c1a28485be70f20c07 : SUCCESS in 9m 52s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/4e4c832d68b34e2888ccb2a008666453 : SUCCESS in 15m 49s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/20df453b78a540a78510643df895c023 : SUCCESS in 1h 01m 03s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/337de86257554a6592bac006f5ec472d : SUCCESS in 4m 29s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/ec646e239a774a5d8af15ace20ff7836 : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a9e3f68097d4c58bf91fd3709d02acb : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2ff9c526dc8475ab59328b31cdc897c : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":1},{"id":"26d23b3019fceb7d163ecc164d6874eadcff0cef","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-21 14:13:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"db1488f50b1d781e5f075ce0eec71467c156a2fa","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-21 14:22:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e31fce02060bc065ff8ed54ff1995f388bc92de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 14:23:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cd4b0db09cb61041973c727e4df6130ad7e15f84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 16:49:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b23ff513840454f880f94e77e52e9e0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f947d3e232a1465480def42cd648927e : SUCCESS in 1h 07m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd3c7445b6f04abe85768cf1dc821efc : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e5990e9d97ce4fb5857206a0abd89ddb : SUCCESS in 8m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d439001cb504cf19c1049e4020c0699 : SUCCESS in 7m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/792b041db7f14430aa635edbf32c6ac5 : SUCCESS in 5m 23s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/764f9bf77c544b27a74281f63395e9a1 : SUCCESS in 49m 47s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ca089c511e2445f793481377d04732a5 : SUCCESS in 3m 13s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/4049bc32ac9847ac833e880063a12228 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/218e7f8e0ae24fdabd487310e481246e : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/755c18b11db749399f5214711ad5676b : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"4fb63aa346ea6cdb9462f73113cec752abc524d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 16:49:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d398c4df338b5a1c6ed43f96a02754701ea452e","revisions":{"5d398c4df338b5a1c6ed43f96a02754701ea452e":{"kind":"REWORK","_number":1,"created":"2026-09-15 09:26:34.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/88/1005688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/88/1005688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/88/1005688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/88/1005688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/88/1005688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/88/1005688/1"}}},"commit":{"parents":[{"commit":"ac0679fc8a611fd2a8984fcc5e6c2474caf9d355","subject":"typing: Remove unnecessary use of TYPE_CHECKING","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/ac0679fc8a611fd2a8984fcc5e6c2474caf9d355"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2026-09-15 09:26:30.000000000","tz":120},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2026-09-15 09:26:30.000000000","tz":120},"subject":"Default OIDC device authorization polling interval to 5s","message":"Default OIDC device authorization polling interval to 5s\n\nRFC 8628 makes the device authorization response interval optional.\nReading it with [\"interval\"] raised KeyError when an IdP omitted the\nfield.\n\nAssisted-by: Claude Sonnet 4.6 \u003cnoreply@anthropic.com\u003e\nChange-Id: Ied2aa68bb461c2166053ef9e38a0cbe856316a26\nSigned-off-by: Grzegorz Grasza \u003cxek@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/5d398c4df338b5a1c6ed43f96a02754701ea452e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/5d398c4df338b5a1c6ed43f96a02754701ea452e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
