)]}'
{"id":"openstack%2Fkeystoneauth~1005694","triplet_id":"openstack%2Fkeystoneauth~master~I1936594aebb8d4db1742cbf8354e49c878be03f3","project":"openstack/keystoneauth","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-15 21:06:49.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I1936594aebb8d4db1742cbf8354e49c878be03f3","subject":"pre-commit: Add codespell","status":"MERGED","created":"2026-09-15 09:56:07.000000000","updated":"2026-09-15 22:14:42.000000000","submitted":"2026-09-15 22:13:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005694","meta_rev_id":"1055062633d461ec95a6ebcd305de1dc0f6a8eec","_number":1005694,"virtual_id_number":1005694,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-15 22:13:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-15 21:06:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-15 21:06:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 11:53:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba102ad55662dd202690c98bef7890236588f5f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 09:56:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6bb1f7a04eb498efa8fa77008d9b8f6ba19d0c69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 11:53:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/243015c8ca784437bd1df772cdd4a5c5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/37ea8eb04da3478c8e20a6eb5217902c : SUCCESS in 3m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1ea9308b61444a66b352874bbb19cb10 : SUCCESS in 2m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da044c05d42b44cc8d229253077f1fdf : SUCCESS in 1h 51m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed86b66caf4841789e9736de5feb6669 : SUCCESS in 10m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c7216b5fb96e49a892e31dfc98349fec : SUCCESS in 7m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d84faea8a9764d4f9bb6f6c88c2c4841 : SUCCESS in 8m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1f00331ccc4b4ca0a0fc9d54dc2d6cff : SUCCESS in 10m 40s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/3c02c5d2de074f2ba4cf05fe538cad6c : SUCCESS in 15m 53s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/37393b7066644976b09c449db31c6e61 : SUCCESS in 58m 14s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f2116f685f284b0b80b67d4cf7da43ce : SUCCESS in 4m 40s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/6bb775501a0545c9925bc01bb8ca5e1f : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32bfecdd3a654e80bf76f23c7f09693f : FAILURE in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f17120ef578147a78f0f7e0abd90bba1 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"5cc2fcfbbdfede89571165b03f50d309ea32b10d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 13:09:01.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"034521dfeab5afccf5f88a30560f4360d4d35575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 14:50:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcce8a9aa7f5450998a844ceab03a1d2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1e1e3261c66947d9a40eaf353a963dee : SUCCESS in 4m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0db03b35b04b442ba7c00302a2c231c5 : SUCCESS in 2m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d8f3dc8b82604724ac4edbf0469aebfa : SUCCESS in 1h 35m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a574aa07d59a4396ad461c12b6a0ed81 : SUCCESS in 8m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/acc9d8d0c46f48648a742b70a01f9919 : SUCCESS in 6m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/820225a513c34803aaa13f322a3dcf2c : SUCCESS in 7m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/efbc574292704cb08f7b15ccac9365cd : SUCCESS in 6m 17s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/16b30386789c40d5b3e82aa4a30b61b5 : SUCCESS in 8m 27s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/bb629004eff245028090c70e9996c53b : SUCCESS in 57m 41s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/14e0a94b06164a20ba31cac5739e89af : SUCCESS in 3m 36s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/1f8e8bc70b7a4fe0bf2a28495400977c : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/965cf0f936064f51b703278f591f6d63 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7efae36afe6d4a309c98aeae1e5405dd : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":2},{"id":"54232265b16ac330d6edaaf54b5f834ab47c9e8b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 21:06:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"12fb0dd7b00ce30b5fb24b8b159cabc54cac4121","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 21:07:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c288df14253299c70547545eb3b9c2262eb86b34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:13:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9233a65c883846f2ad60605d9ca1ca23\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/70c0aecbdf004e2989957afd98ff475f : SUCCESS in 4m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7dd86b70af624cbeaa311f9f81e58bd1 : SUCCESS in 3m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/51b88d5634d74a0fb0d7d5b253d6ea5a : SUCCESS in 55m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/699b5d2fc8c44d1fac39cf696fcf4461 : SUCCESS in 8m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b706fe8cfb1d4e81b45f019f6db4e93a : SUCCESS in 7m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/760ad2782bdc4adc86cabc10c12ebd9a : SUCCESS in 7m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2bad11612fb94d5ca0c2c41a7e54dc8f : SUCCESS in 12m 57s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/38c4d79204fb4c1fa89081d6a89d0017 : SUCCESS in 1h 00m 19s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/fabc047978324bf38cf690eb8b222c70 : SUCCESS in 2m 41s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/abede14c254c40078405be5c96ad912f : SUCCESS in 10m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4a5b5ca20364d5fa199d71992f00864 : SUCCESS in 9m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6f21e461fd14b71808a5a66678336a3 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"33bdb37997eeea38c767e840773344e051981b80","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:13:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1055062633d461ec95a6ebcd305de1dc0f6a8eec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:14:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4230310804d549db9b104305ea56067e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd6f0ce8fddd411bab5eec1d2d12af4f : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8dfb8618886b468697ce2e0efc3ecdbb : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d9510af99bb7c5479da4eb316e459437d770224e","revisions":{"c4a4ae730e45dacec2f91f46bae0eaa01044e183":{"kind":"REWORK","_number":1,"created":"2026-09-15 09:56:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/94/1005694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/94/1005694/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/94/1005694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/94/1005694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/94/1005694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/94/1005694/1"}}},"commit":{"parents":[{"commit":"b3b61a34eeef9c6114dd3476a0682d26adf5db86","subject":"trivial: Remove type info from docstrings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/b3b61a34eeef9c6114dd3476a0682d26adf5db86"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 09:43:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 09:51:27.000000000","tz":60},"subject":"pre-commit: Add codespell","message":"pre-commit: Add codespell\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I1936594aebb8d4db1742cbf8354e49c878be03f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/c4a4ae730e45dacec2f91f46bae0eaa01044e183"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/c4a4ae730e45dacec2f91f46bae0eaa01044e183"}]},"branch":"refs/heads/master"},"d9510af99bb7c5479da4eb316e459437d770224e":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-15 13:09:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/94/1005694/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/94/1005694/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/94/1005694/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/94/1005694/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/94/1005694/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/94/1005694/2"}}},"commit":{"parents":[{"commit":"8980efafacd178f72e5e407e29743872280e5667","subject":"trivial: Remove type info from docstrings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/8980efafacd178f72e5e407e29743872280e5667"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 09:43:05.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 13:06:33.000000000","tz":60},"subject":"pre-commit: Add codespell","message":"pre-commit: Add codespell\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I1936594aebb8d4db1742cbf8354e49c878be03f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/d9510af99bb7c5479da4eb316e459437d770224e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/d9510af99bb7c5479da4eb316e459437d770224e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
