)]}'
{"id":"openstack%2Fkeystoneauth~1005695","triplet_id":"openstack%2Fkeystoneauth~master~I42b7cd16300fc0b56c61131226896a5ef6262387","project":"openstack/keystoneauth","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-15 21:06:54.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I42b7cd16300fc0b56c61131226896a5ef6262387","subject":"trivial: Rewrap opt help strings","status":"MERGED","created":"2026-09-15 09:56:07.000000000","updated":"2026-09-15 22:59:03.000000000","submitted":"2026-09-15 22:57:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"1005695","meta_rev_id":"fdeb7efff877226b4d064d347600e230e829becf","_number":1005695,"virtual_id_number":1005695,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-15 22:57:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-09-15 21:06:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-09-15 21:06:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-15 11:27:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"77fcee5269ca0a5c6bdafb8ad93157b86f6687fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 09:56:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b4d0043655d1435ba221701e8f7fe0b7e6fa073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 11:27:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/058857daaf86406a918e5270784e33f4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7499881b28af46b798374a75fca2393b : SUCCESS in 1h 29m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86d86945d2cd4c519712b5cef685ef41 : SUCCESS in 7m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f32510bf4a9447f3bfccac6db2682b4d : SUCCESS in 6m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8854e9829bca47ba8b06be57e6e7fb77 : SUCCESS in 7m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a3b4b6666d0f47069b68c34eb6c2632a : SUCCESS in 10m 27s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/f6d89a85ae02497a84aef75ac2ce4f90 : SUCCESS in 14m 37s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/64f4795d9c084521918aaa51b5fcddfc : SUCCESS in 1h 00m 29s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f0e9b76e2867488faf95ad071abc1cd7 : SUCCESS in 5m 28s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/7f869a04b4744b4b8ea0235d588f7b5c : SUCCESS in 9m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52995b94a54f4718b380f3c107154853 : FAILURE in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/120138f08bb9400eb5d88389791c7740 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"abe1dbde3eb98120f5670ca66facbb5da9837e81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 13:09:01.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"d6d1f4544f8c6cf156f2c7626ef12b1971613eb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 14:18:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/027cf44131ca4b4b89dadf56cc65838d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3806add61767412d954fd93da7ab9b08 : SUCCESS in 42m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e70846c649644dddaac68964ccdba71b : SUCCESS in 8m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f1d5d0c3313244548d2685a6ef42dd63 : SUCCESS in 7m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0430cd44a50f4b8ea799717eb77be4e7 : SUCCESS in 7m 17s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7ef21b5f3ded43e0ba198ce8e82da430 : SUCCESS in 9m 56s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/b3ccd27dc4a749e2898f4509cb703b93 : SUCCESS in 14m 37s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d8c8bd8d0d8d46bea077881252c0c577 : SUCCESS in 1h 03m 44s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/90fd157113af4080b80fb9ba89606151 : SUCCESS in 4m 46s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/34237732564846c281dd6a4ebc3149a1 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4815cf17f8354a1f9f1f00e62b5fa839 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/288ff9ba71434ae2825fdfd2e2896ae7 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":2},{"id":"5fe0694b421b5ad86cbb1afaf6b8b5faa7086f24","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-15 21:06:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6a7ec7320c0735613a7965ca01dd2b5ec656be98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 21:07:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b5cca11fc87372c10cd201b8ad18aae9f32bf0a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:57:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc79dbf2c195478bb51561a8c44caae9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/efd9697c65b64dc1be14d5e72484d13d : SUCCESS in 1h 44m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2c638b596b54eddba46d09b679a4a88 : SUCCESS in 8m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/682fc3b525c54c60beae71b350339b99 : SUCCESS in 3m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/be4e9e5dd28a40a883eb8d1f3ae65150 : SUCCESS in 3m 55s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5760eec8e4924458a7ca78c99ea3a653 : SUCCESS in 10m 46s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/71e25bbb64364d14bd30c2e5f1b97dc7 : SUCCESS in 49m 42s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/be98c75475b84da59709bb713256c5fb : SUCCESS in 2m 44s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/54b151042d234d09a9c6c1b01fb4ef75 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4b1b077997c45d28a7c6787fcc6b851 : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/210f08a4cad84cd08a658da3e446cd4e : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":2},{"id":"adfd3f498750087e3c27196a54eb3fa0e1c1078f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:57:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fdeb7efff877226b4d064d347600e230e829becf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 22:59:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d70f425a04414d3a82e0c7a28117254c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92d85f1e54064f9bb6498d3495082e9b : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3d9ae0c7ede49d5732f5c02f48c58ffec20400c1","revisions":{"c2b323deb016af5cae46510886f72d482c0303d6":{"kind":"REWORK","_number":1,"created":"2026-09-15 09:56:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/95/1005695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/95/1005695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/95/1005695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/95/1005695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/95/1005695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/95/1005695/1"}}},"commit":{"parents":[{"commit":"c4a4ae730e45dacec2f91f46bae0eaa01044e183","subject":"pre-commit: Add codespell","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/c4a4ae730e45dacec2f91f46bae0eaa01044e183"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 09:55:57.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 09:55:57.000000000","tz":60},"subject":"trivial: Rewrap opt help strings","message":"trivial: Rewrap opt help strings\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I42b7cd16300fc0b56c61131226896a5ef6262387\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/c2b323deb016af5cae46510886f72d482c0303d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/c2b323deb016af5cae46510886f72d482c0303d6"}]},"branch":"refs/heads/master"},"3d9ae0c7ede49d5732f5c02f48c58ffec20400c1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-09-15 13:09:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/95/1005695/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/95/1005695/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/95/1005695/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/95/1005695/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/95/1005695/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/95/1005695/2"}}},"commit":{"parents":[{"commit":"d9510af99bb7c5479da4eb316e459437d770224e","subject":"pre-commit: Add codespell","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/d9510af99bb7c5479da4eb316e459437d770224e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 09:55:57.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-15 13:06:59.000000000","tz":60},"subject":"trivial: Rewrap opt help strings","message":"trivial: Rewrap opt help strings\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I42b7cd16300fc0b56c61131226896a5ef6262387\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/3d9ae0c7ede49d5732f5c02f48c58ffec20400c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/3d9ae0c7ede49d5732f5c02f48c58ffec20400c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
