)]}'
{"id":"openstack%2Fkeystoneauth~549536","triplet_id":"openstack%2Fkeystoneauth~master~I02b333b1cf75419bfad64159976f5aba0f304125","project":"openstack/keystoneauth","branch":"master","topic":"openstack/requirements","hashtags":[],"change_id":"I02b333b1cf75419bfad64159976f5aba0f304125","subject":"Updated from global requirements","status":"MERGED","created":"2018-03-04 10:06:27.000000000","updated":"2018-03-18 17:48:45.000000000","submitted":"2018-03-18 17:48:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"549536-1521395325601-3f670796","meta_rev_id":"f5034728340cb4c948ba17b8eddb0249129257ec","_number":549536,"virtual_id_number":549536,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":2,"date":"2018-03-18 17:48:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":2,"date":"2018-03-18 16:29:27.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"all":[{"value":1,"date":"2018-03-18 16:29:27.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-03-18 16:29:27.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2018-03-18 17:48:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6f597cdea5e853fd71900e13d029bc7fd3d918a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2018-03-04 10:06:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d44a4917ab93c0dcb658a7131a683d0935805fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-04 11:23:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/36/549536/1/check/openstack-tox-pep8/625c78b/ : SUCCESS in 3m 01s\n- openstack-tox-py27 http://logs.openstack.org/36/549536/1/check/openstack-tox-py27/2afa3ce/ : SUCCESS in 3m 04s\n- openstack-tox-py35 http://logs.openstack.org/36/549536/1/check/openstack-tox-py35/6227874/ : SUCCESS in 2m 43s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/549536/1/check/build-openstack-sphinx-docs/63af68f/html/ : SUCCESS in 3m 32s\n- requirements-check http://logs.openstack.org/36/549536/1/check/requirements-check/4ac2ebf/ : SUCCESS in 1m 25s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/36/549536/1/check/legacy-tempest-dsvm-neutron-src/bac285f/ : SUCCESS in 1h 14m 28s\n- build-openstack-releasenotes http://logs.openstack.org/36/549536/1/check/build-openstack-releasenotes/1c6bd5a/html/ : SUCCESS in 2m 18s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/36/549536/1/check/openstacksdk-functional-devstack-tips/888d9fb/ : SUCCESS in 55m 41s\n- openstacksdk-functional-devstack-tips-python3 http://logs.openstack.org/36/549536/1/check/openstacksdk-functional-devstack-tips-python3/db251a7/ : SUCCESS in 59m 15s\n- openstacksdk-tox-py27-tips http://logs.openstack.org/36/549536/1/check/openstacksdk-tox-py27-tips/6b29804/ : SUCCESS in 2m 55s\n- openstacksdk-tox-py35-tips http://logs.openstack.org/36/549536/1/check/openstacksdk-tox-py35-tips/829a258/ : SUCCESS in 2m 59s\n- osc-tox-py27-tips http://logs.openstack.org/36/549536/1/check/osc-tox-py27-tips/97541b2/ : SUCCESS in 5m 49s\n- osc-tox-py35-tips http://logs.openstack.org/36/549536/1/check/osc-tox-py35-tips/464d2b1/ : SUCCESS in 4m 07s\n- shade-functional-devstack-tips http://logs.openstack.org/36/549536/1/check/shade-functional-devstack-tips/4014181/ : SUCCESS in 35m 17s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/36/549536/1/check/shade-functional-devstack-tips-python3/f17aea1/ : SUCCESS in 35m 39s\n- shade-tox-py27-tips http://logs.openstack.org/36/549536/1/check/shade-tox-py27-tips/a222b49/ : SUCCESS in 2m 45s\n- shade-tox-py35-tips http://logs.openstack.org/36/549536/1/check/shade-tox-py35-tips/b1b9c68/ : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":1},{"id":"8da9165105c2302cf2dbe438f431fcbd2e8b5239","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2018-03-10 13:17:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"13942308e55ab4f86bc97b6c199bf316f4bed4e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-10 14:51:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/36/549536/2/check/openstack-tox-pep8/b8de47c/ : SUCCESS in 2m 54s\n- openstack-tox-py27 http://logs.openstack.org/36/549536/2/check/openstack-tox-py27/983b3c4/ : SUCCESS in 3m 03s\n- openstack-tox-py35 http://logs.openstack.org/36/549536/2/check/openstack-tox-py35/8470f20/ : SUCCESS in 2m 48s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/549536/2/check/build-openstack-sphinx-docs/06e1878/html/ : SUCCESS in 3m 40s\n- requirements-check http://logs.openstack.org/36/549536/2/check/requirements-check/ba5c314/ : SUCCESS in 1m 52s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/36/549536/2/check/legacy-tempest-dsvm-neutron-src/f191d76/ : SUCCESS in 1h 31m 41s\n- build-openstack-releasenotes http://logs.openstack.org/36/549536/2/check/build-openstack-releasenotes/c6af2c3/html/ : SUCCESS in 2m 07s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/36/549536/2/check/openstacksdk-functional-devstack-tips/a201db7/ : FAILURE in 1h 25m 19s\n- openstacksdk-functional-devstack-tips-python3 http://logs.openstack.org/36/549536/2/check/openstacksdk-functional-devstack-tips-python3/1dde875/ : SUCCESS in 52m 51s\n- openstacksdk-tox-py27-tips http://logs.openstack.org/36/549536/2/check/openstacksdk-tox-py27-tips/eb38e03/ : SUCCESS in 3m 23s\n- openstacksdk-tox-py35-tips http://logs.openstack.org/36/549536/2/check/openstacksdk-tox-py35-tips/34e6be0/ : SUCCESS in 3m 44s\n- osc-tox-py27-tips http://logs.openstack.org/36/549536/2/check/osc-tox-py27-tips/ee09672/ : SUCCESS in 5m 30s\n- osc-tox-py35-tips http://logs.openstack.org/36/549536/2/check/osc-tox-py35-tips/f60bbe7/ : SUCCESS in 5m 07s\n- shade-functional-devstack-tips http://logs.openstack.org/36/549536/2/check/shade-functional-devstack-tips/bf4abf2/ : SUCCESS in 37m 30s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/36/549536/2/check/shade-functional-devstack-tips-python3/7434324/ : SUCCESS in 36m 16s\n- shade-tox-py27-tips http://logs.openstack.org/36/549536/2/check/shade-tox-py27-tips/813e01a/ : SUCCESS in 2m 23s\n- shade-tox-py35-tips http://logs.openstack.org/36/549536/2/check/shade-tox-py35-tips/14b3d6f/ : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"8b76f584d7fbcea0df8d6f510e9e73b41328c3a5","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2018-03-13 06:56:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aeff20f4796ff93ae9d36b384b73c09fcf6d80c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-13 08:20:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/36/549536/3/check/openstack-tox-pep8/481db30/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/36/549536/3/check/openstack-tox-py27/882960a/ : SUCCESS in 3m 10s\n- openstack-tox-py35 http://logs.openstack.org/36/549536/3/check/openstack-tox-py35/ee93a56/ : SUCCESS in 3m 16s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/549536/3/check/build-openstack-sphinx-docs/ec8a945/ : FAILURE in 3m 01s\n- requirements-check http://logs.openstack.org/36/549536/3/check/requirements-check/c67ddbb/ : SUCCESS in 2m 04s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/36/549536/3/check/legacy-tempest-dsvm-neutron-src/0fdecc4/ : SUCCESS in 1h 22m 03s\n- build-openstack-releasenotes http://logs.openstack.org/36/549536/3/check/build-openstack-releasenotes/b1e2146/html/ : SUCCESS in 2m 16s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/36/549536/3/check/openstacksdk-functional-devstack-tips/f0445a7/ : FAILURE in 57m 08s\n- openstacksdk-functional-devstack-tips-python3 http://logs.openstack.org/36/549536/3/check/openstacksdk-functional-devstack-tips-python3/b0b7acb/ : FAILURE in 55m 06s\n- openstacksdk-tox-py27-tips http://logs.openstack.org/36/549536/3/check/openstacksdk-tox-py27-tips/37af72e/ : SUCCESS in 3m 03s\n- openstacksdk-tox-py35-tips http://logs.openstack.org/36/549536/3/check/openstacksdk-tox-py35-tips/c15d7fc/ : SUCCESS in 3m 29s\n- osc-tox-py27-tips http://logs.openstack.org/36/549536/3/check/osc-tox-py27-tips/cfd884f/ : SUCCESS in 6m 36s\n- osc-tox-py35-tips http://logs.openstack.org/36/549536/3/check/osc-tox-py35-tips/93f2042/ : SUCCESS in 4m 48s\n- shade-functional-devstack-tips http://logs.openstack.org/36/549536/3/check/shade-functional-devstack-tips/7994ccb/ : SUCCESS in 36m 42s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/36/549536/3/check/shade-functional-devstack-tips-python3/7a0bb27/ : SUCCESS in 34m 09s\n- shade-tox-py27-tips http://logs.openstack.org/36/549536/3/check/shade-tox-py27-tips/99a453f/ : SUCCESS in 3m 31s\n- shade-tox-py35-tips http://logs.openstack.org/36/549536/3/check/shade-tox-py35-tips/b3e46ef/ : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":3},{"id":"b3a9dc284c94cafa03cd28ef910eabdb8272a0b2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2018-03-14 05:33:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"296d0967ead807b870cd455fdf70bfd49c51f651","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-14 07:10:37.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/36/549536/4/check/openstack-tox-pep8/d69901c/ : SUCCESS in 2m 33s\n- openstack-tox-py27 http://logs.openstack.org/36/549536/4/check/openstack-tox-py27/3ed9274/ : SUCCESS in 3m 44s\n- openstack-tox-py35 http://logs.openstack.org/36/549536/4/check/openstack-tox-py35/51ae820/ : SUCCESS in 3m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/549536/4/check/build-openstack-sphinx-docs/6c34e97/ : FAILURE in 3m 54s\n- requirements-check http://logs.openstack.org/36/549536/4/check/requirements-check/851dc08/ : SUCCESS in 1m 51s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/36/549536/4/check/legacy-tempest-dsvm-neutron-src/dd7c1a2/ : SUCCESS in 1h 29m 10s\n- build-openstack-releasenotes http://logs.openstack.org/36/549536/4/check/build-openstack-releasenotes/c88de41/html/ : SUCCESS in 2m 49s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/36/549536/4/check/openstacksdk-functional-devstack-tips/da01bb8/ : SUCCESS in 59m 37s\n- openstacksdk-functional-devstack-tips-python3 http://logs.openstack.org/36/549536/4/check/openstacksdk-functional-devstack-tips-python3/2c05f68/ : SUCCESS in 59m 28s\n- openstacksdk-tox-py27-tips http://logs.openstack.org/36/549536/4/check/openstacksdk-tox-py27-tips/fa9f15b/ : SUCCESS in 3m 29s\n- openstacksdk-tox-py35-tips http://logs.openstack.org/36/549536/4/check/openstacksdk-tox-py35-tips/58bc09f/ : SUCCESS in 3m 30s\n- osc-tox-py27-tips http://logs.openstack.org/36/549536/4/check/osc-tox-py27-tips/d323f81/ : SUCCESS in 6m 44s\n- osc-tox-py35-tips http://logs.openstack.org/36/549536/4/check/osc-tox-py35-tips/e3771a3/ : SUCCESS in 5m 08s\n- shade-functional-devstack-tips http://logs.openstack.org/36/549536/4/check/shade-functional-devstack-tips/547a599/ : SUCCESS in 44m 13s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/36/549536/4/check/shade-functional-devstack-tips-python3/ca2f654/ : SUCCESS in 32m 59s\n- shade-tox-py27-tips http://logs.openstack.org/36/549536/4/check/shade-tox-py27-tips/a76adba/ : SUCCESS in 3m 16s\n- shade-tox-py35-tips http://logs.openstack.org/36/549536/4/check/shade-tox-py35-tips/b9ff180/ : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":4},{"id":"d85663faf5727ac8cc777d2d1ee610859bd94fb7","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2018-03-15 07:03:38.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"625014532b8be2554d7a46c7f7930b46d1131a65","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-15 08:16:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/36/549536/5/check/openstack-tox-pep8/7a879be/ : SUCCESS in 3m 18s\n- openstack-tox-py27 http://logs.openstack.org/36/549536/5/check/openstack-tox-py27/0269a2c/ : SUCCESS in 3m 02s\n- openstack-tox-py35 http://logs.openstack.org/36/549536/5/check/openstack-tox-py35/3b22b72/ : SUCCESS in 2m 43s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/549536/5/check/build-openstack-sphinx-docs/3bc6720/html/ : SUCCESS in 3m 18s\n- requirements-check http://logs.openstack.org/36/549536/5/check/requirements-check/405609e/ : SUCCESS in 1m 44s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/36/549536/5/check/legacy-tempest-dsvm-neutron-src/58dfcad/ : SUCCESS in 1h 10m 38s\n- build-openstack-releasenotes http://logs.openstack.org/36/549536/5/check/build-openstack-releasenotes/f3c2010/html/ : SUCCESS in 2m 09s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/36/549536/5/check/openstacksdk-functional-devstack-tips/6d01c03/ : POST_FAILURE in 51m 02s\n- openstacksdk-functional-devstack-tips-python3 http://logs.openstack.org/36/549536/5/check/openstacksdk-functional-devstack-tips-python3/b8e3f9e/ : POST_FAILURE in 50m 57s\n- openstacksdk-tox-py27-tips http://logs.openstack.org/36/549536/5/check/openstacksdk-tox-py27-tips/2cb3813/ : SUCCESS in 3m 16s\n- openstacksdk-tox-py35-tips http://logs.openstack.org/36/549536/5/check/openstacksdk-tox-py35-tips/7f548ff/ : SUCCESS in 3m 20s\n- osc-tox-py27-tips http://logs.openstack.org/36/549536/5/check/osc-tox-py27-tips/0e1483d/ : SUCCESS in 5m 29s\n- osc-tox-py35-tips http://logs.openstack.org/36/549536/5/check/osc-tox-py35-tips/61b7d20/ : SUCCESS in 5m 33s\n- shade-functional-devstack-tips http://logs.openstack.org/36/549536/5/check/shade-functional-devstack-tips/9635495/ : POST_FAILURE in 32m 41s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/36/549536/5/check/shade-functional-devstack-tips-python3/e8e4c7b/ : POST_FAILURE in 36m 53s\n- shade-tox-py27-tips http://logs.openstack.org/36/549536/5/check/shade-tox-py27-tips/fcc97ba/ : SUCCESS in 2m 46s\n- shade-tox-py35-tips http://logs.openstack.org/36/549536/5/check/shade-tox-py35-tips/3a441e4/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":5},{"id":"93c9013adfd9da90ac8804a22c07aed818f0b71c","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2018-03-17 23:42:55.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"01e16ba79bfa780d492f62dc991995c8cd65193b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-18 01:00:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/36/549536/5/check/openstack-tox-pep8/5d55430/ : SUCCESS in 2m 40s\n- openstack-tox-py27 http://logs.openstack.org/36/549536/5/check/openstack-tox-py27/07a62e7/ : SUCCESS in 3m 28s\n- openstack-tox-py35 http://logs.openstack.org/36/549536/5/check/openstack-tox-py35/dd84d50/ : SUCCESS in 2m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/549536/5/check/build-openstack-sphinx-docs/2074010/html/ : SUCCESS in 2m 48s\n- requirements-check http://logs.openstack.org/36/549536/5/check/requirements-check/891f39f/ : SUCCESS in 1m 36s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/36/549536/5/check/legacy-tempest-dsvm-neutron-src/14dcf7b/ : SUCCESS in 1h 14m 03s\n- build-openstack-releasenotes http://logs.openstack.org/36/549536/5/check/build-openstack-releasenotes/4e54aa8/html/ : SUCCESS in 1m 55s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/36/549536/5/check/openstacksdk-functional-devstack-tips/e7b5248/ : SUCCESS in 1h 00m 36s\n- openstacksdk-functional-devstack-tips-python3 http://logs.openstack.org/36/549536/5/check/openstacksdk-functional-devstack-tips-python3/a7d733b/ : SUCCESS in 56m 12s\n- openstacksdk-tox-py27-tips http://logs.openstack.org/36/549536/5/check/openstacksdk-tox-py27-tips/9910c29/ : SUCCESS in 2m 54s\n- openstacksdk-tox-py35-tips http://logs.openstack.org/36/549536/5/check/openstacksdk-tox-py35-tips/e32e6b9/ : SUCCESS in 2m 47s\n- osc-tox-py27-tips http://logs.openstack.org/36/549536/5/check/osc-tox-py27-tips/4466d82/ : SUCCESS in 4m 59s\n- osc-tox-py35-tips http://logs.openstack.org/36/549536/5/check/osc-tox-py35-tips/af80bfd/ : SUCCESS in 5m 43s\n- shade-functional-devstack-tips http://logs.openstack.org/36/549536/5/check/shade-functional-devstack-tips/5019b6a/ : SUCCESS in 32m 13s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/36/549536/5/check/shade-functional-devstack-tips-python3/386ae10/ : SUCCESS in 38m 25s\n- shade-tox-py27-tips http://logs.openstack.org/36/549536/5/check/shade-tox-py27-tips/f5a833a/ : SUCCESS in 2m 33s\n- shade-tox-py35-tips http://logs.openstack.org/36/549536/5/check/shade-tox-py35-tips/c01f8be/ : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":5},{"id":"520ca9404b9fc4bf39448c2efa7d1c8f88fc6852","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2018-03-18 16:29:27.000000000","message":"Patch Set 5: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1e56e415b7d69b7ca95dfaa2175c16cd5ef6c094","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-18 16:29:38.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"54de5ed3d129ea7971aa717af93d87a224f1fa55","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-18 17:48:45.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/36/549536/5/gate/openstack-tox-pep8/4497ee4/ : SUCCESS in 2m 37s\n- openstack-tox-py27 http://logs.openstack.org/36/549536/5/gate/openstack-tox-py27/c7d8c8d/ : SUCCESS in 3m 14s\n- openstack-tox-py35 http://logs.openstack.org/36/549536/5/gate/openstack-tox-py35/7915820/ : SUCCESS in 3m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/36/549536/5/gate/build-openstack-sphinx-docs/ad903d2/html/ : SUCCESS in 2m 58s\n- requirements-check http://logs.openstack.org/36/549536/5/gate/requirements-check/f4ce751/ : SUCCESS in 1m 46s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/36/549536/5/gate/legacy-tempest-dsvm-neutron-src/fa7ed8b/ : SUCCESS in 1h 17m 54s\n- build-openstack-releasenotes http://logs.openstack.org/36/549536/5/gate/build-openstack-releasenotes/244e62e/html/ : SUCCESS in 2m 14s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/36/549536/5/gate/openstacksdk-functional-devstack-tips/0908465/ : SUCCESS in 1h 01m 34s\n- openstacksdk-functional-devstack-tips-python3 http://logs.openstack.org/36/549536/5/gate/openstacksdk-functional-devstack-tips-python3/dc716b0/ : SUCCESS in 55m 27s\n- openstacksdk-tox-py27-tips http://logs.openstack.org/36/549536/5/gate/openstacksdk-tox-py27-tips/ebb4020/ : SUCCESS in 2m 32s\n- openstacksdk-tox-py35-tips http://logs.openstack.org/36/549536/5/gate/openstacksdk-tox-py35-tips/86da973/ : SUCCESS in 3m 15s\n- osc-tox-py27-tips http://logs.openstack.org/36/549536/5/gate/osc-tox-py27-tips/ef2cd88/ : SUCCESS in 4m 03s\n- osc-tox-py35-tips http://logs.openstack.org/36/549536/5/gate/osc-tox-py35-tips/3224da8/ : SUCCESS in 6m 01s\n- shade-functional-devstack-tips http://logs.openstack.org/36/549536/5/gate/shade-functional-devstack-tips/508e106/ : SUCCESS in 40m 03s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/36/549536/5/gate/shade-functional-devstack-tips-python3/5ef9049/ : SUCCESS in 40m 25s\n- shade-tox-py27-tips http://logs.openstack.org/36/549536/5/gate/shade-tox-py27-tips/a34a716/ : SUCCESS in 3m 06s\n- shade-tox-py35-tips http://logs.openstack.org/36/549536/5/gate/shade-tox-py35-tips/860ca46/ : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":5},{"id":"b7f17e17a528b40ed7e617c0c3fc090eb3dd22ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-18 17:48:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ce78397bffa61eb52ca7fe1b90b35d6c3cb2c1b1","revisions":{"306fa82b6c9bf234e41439505f66b48d0a20794e":{"kind":"REWORK","_number":1,"created":"2018-03-04 10:06:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/549536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/36/549536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/1"}}},"commit":{"parents":[{"commit":"80bfea8b3ba9955245fca91b69db855f71ab277a","subject":"Merge \"Override support message in AuthorizationFailure  class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/80bfea8b3ba9955245fca91b69db855f71ab277a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-04 10:06:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-04 10:06:25.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I02b333b1cf75419bfad64159976f5aba0f304125\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/306fa82b6c9bf234e41439505f66b48d0a20794e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/306fa82b6c9bf234e41439505f66b48d0a20794e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"80bfea8b3ba9955245fca91b69db855f71ab277a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"6c6b04759913e1247fdbdd5d6cb176dc1f964bdf":{"kind":"REWORK","_number":2,"created":"2018-03-10 13:17:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/549536/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/36/549536/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/2"}}},"commit":{"parents":[{"commit":"80bfea8b3ba9955245fca91b69db855f71ab277a","subject":"Merge \"Override support message in AuthorizationFailure  class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/80bfea8b3ba9955245fca91b69db855f71ab277a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-10 13:17:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-10 13:17:02.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I02b333b1cf75419bfad64159976f5aba0f304125\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/6c6b04759913e1247fdbdd5d6cb176dc1f964bdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/6c6b04759913e1247fdbdd5d6cb176dc1f964bdf"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"80bfea8b3ba9955245fca91b69db855f71ab277a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"6e554ddd62c48e00a31294a20e8ab5df12726e3d":{"kind":"REWORK","_number":3,"created":"2018-03-13 06:56:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/549536/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/36/549536/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/3"}}},"commit":{"parents":[{"commit":"80bfea8b3ba9955245fca91b69db855f71ab277a","subject":"Merge \"Override support message in AuthorizationFailure  class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/80bfea8b3ba9955245fca91b69db855f71ab277a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-13 06:56:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-13 06:56:27.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I02b333b1cf75419bfad64159976f5aba0f304125\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/6e554ddd62c48e00a31294a20e8ab5df12726e3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/6e554ddd62c48e00a31294a20e8ab5df12726e3d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"80bfea8b3ba9955245fca91b69db855f71ab277a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"66be1cf8ac724b678df6d0bed296d52237b2d227":{"kind":"REWORK","_number":4,"created":"2018-03-14 05:33:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/549536/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/36/549536/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/4"}}},"commit":{"parents":[{"commit":"80bfea8b3ba9955245fca91b69db855f71ab277a","subject":"Merge \"Override support message in AuthorizationFailure  class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/80bfea8b3ba9955245fca91b69db855f71ab277a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-14 05:33:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-14 05:33:18.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I02b333b1cf75419bfad64159976f5aba0f304125\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/66be1cf8ac724b678df6d0bed296d52237b2d227"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/66be1cf8ac724b678df6d0bed296d52237b2d227"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"80bfea8b3ba9955245fca91b69db855f71ab277a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"ce78397bffa61eb52ca7fe1b90b35d6c3cb2c1b1":{"kind":"REWORK","_number":5,"created":"2018-03-15 07:03:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/36/549536/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/36/549536/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/36/549536/5"}}},"commit":{"parents":[{"commit":"80bfea8b3ba9955245fca91b69db855f71ab277a","subject":"Merge \"Override support message in AuthorizationFailure  class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/80bfea8b3ba9955245fca91b69db855f71ab277a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-15 07:03:34.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2018-03-15 07:03:34.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I02b333b1cf75419bfad64159976f5aba0f304125\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/ce78397bffa61eb52ca7fe1b90b35d6c3cb2c1b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/ce78397bffa61eb52ca7fe1b90b35d6c3cb2c1b1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"80bfea8b3ba9955245fca91b69db855f71ab277a","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
