)]}'
{"id":"openstack%2Fkeystoneauth~563227","triplet_id":"openstack%2Fkeystoneauth~master~Ib73611ba0da839d46c845967aacf889ccbde9a19","project":"openstack/keystoneauth","branch":"master","topic":"pypi-url","hashtags":[],"change_id":"Ib73611ba0da839d46c845967aacf889ccbde9a19","subject":"Trivial: Update pypi url to new url","status":"ABANDONED","created":"2018-04-20 18:50:48.000000000","updated":"2018-05-15 17:43:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"889125c1a7f54b2110187a534a548dc06fa61e13","_number":563227,"virtual_id_number":563227,"owner":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2018-04-20 18:56:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-04-20 18:56:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"688a88395f8573b8b82b3ca14333f22bd5841435","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2018-04-20 18:50:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ca5fa357aeccb1fc02c2e499beacb3b23b6dfb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-04-20 18:56:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/27/563227/1/check/openstack-tox-pep8/790cac2/ : SUCCESS in 3m 08s\n- build-openstack-sphinx-docs http://logs.openstack.org/27/563227/1/check/build-openstack-sphinx-docs/75564f6/html/ : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":1},{"id":"48e1912f0295f3e7ac32ff9b10df68647a171c8d","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2018-05-15 17:43:27.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8ba350b2559425367ba04359716debddb553034","revisions":{"a8ba350b2559425367ba04359716debddb553034":{"kind":"REWORK","_number":1,"created":"2018-04-20 18:50:48.000000000","uploader":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"ref":"refs/changes/27/563227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/27/563227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/27/563227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/27/563227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/27/563227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/27/563227/1"}}},"commit":{"parents":[{"commit":"f41e30484902d305bab56c9194b32d215f08b7cf","subject":"add lower-constraints job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/f41e30484902d305bab56c9194b32d215f08b7cf"}]}],"author":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2018-04-20 18:50:07.000000000","tz":480},"committer":{"name":"melissaml","email":"ma.lei@99cloud.net","date":"2018-04-20 18:50:41.000000000","tz":480},"subject":"Trivial: Update pypi url to new url","message":"Trivial: Update pypi url to new url\n\nPypi url changed from [1] to [2]\n\n[1] https://pypi.python.org/pypi/\u003cpackage\u003e\n[2] https://pypi.org/project/\u003cpackage\u003e\n\nChange-Id: Ib73611ba0da839d46c845967aacf889ccbde9a19\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/a8ba350b2559425367ba04359716debddb553034"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/a8ba350b2559425367ba04359716debddb553034"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f41e30484902d305bab56c9194b32d215f08b7cf","is_merged_in_target_branch":true,"change_id":"Ie5da1d132321b2cffea3eb8522462daad45900b5","change_number":555625,"patch_set_number":11,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
