)]}'
{"id":"openstack%2Fkeystoneauth~597656","triplet_id":"openstack%2Fkeystoneauth~master~I30f3bc0f5b68190f72b015ae5a14251c91eeee42","project":"openstack/keystoneauth","branch":"master","topic":"python3-first","hashtags":[],"change_id":"I30f3bc0f5b68190f72b015ae5a14251c91eeee42","subject":"switch documentation job to new PTI","status":"MERGED","created":"2018-08-29 20:10:53.000000000","updated":"2018-09-05 21:43:03.000000000","submitted":"2018-09-05 21:43:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"597656-1536183783376-f2fc24b1","meta_rev_id":"846ef5a469702fd9be43001e2b7644bd334296fe","_number":597656,"virtual_id_number":597656,"owner":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2018-08-31 07:38:37.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":2,"date":"2018-09-05 21:43:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2018-09-03 12:18:10.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2018-09-05 20:14:50.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2018-08-30 18:18:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2018-09-01 02:59:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2018-09-05 20:14:50.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-08-30 18:18:46.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2018-08-31 07:38:37.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2018-09-01 02:59:58.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2018-09-03 12:18:10.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2018-09-05 20:14:50.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2018-09-05 21:43:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c5d2aa6bc66b9a6bfe5e034d80f0e971deb0f119","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-08-29 20:10:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e17b71f64ed4e94f73b4fa4988aee0e2399e9dc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-29 22:04:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/56/597656/1/check/openstack-tox-pep8/d51be8b/ : SUCCESS in 4m 24s\n- openstack-tox-py27 http://logs.openstack.org/56/597656/1/check/openstack-tox-py27/2822475/ : SUCCESS in 4m 40s\n- openstack-tox-py35 http://logs.openstack.org/56/597656/1/check/openstack-tox-py35/ec179f0/ : SUCCESS in 4m 44s\n- build-openstack-sphinx-docs http://logs.openstack.org/56/597656/1/check/build-openstack-sphinx-docs/52993db/html/ : SUCCESS in 3m 57s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/56/597656/1/check/legacy-tempest-dsvm-neutron-src/f5991cf/ : SUCCESS in 1h 43m 44s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/56/597656/1/check/openstacksdk-functional-devstack-tips/ce61a21/ : FAILURE in 1h 24m 25s\n- openstacksdk-functional-devstack-tips-python3 http://logs.openstack.org/56/597656/1/check/openstacksdk-functional-devstack-tips-python3/38d1cb4/ : SUCCESS in 1h 16m 46s\n- openstacksdk-tox-py27-tips http://logs.openstack.org/56/597656/1/check/openstacksdk-tox-py27-tips/d490b41/ : SUCCESS in 4m 01s\n- openstacksdk-tox-py35-tips http://logs.openstack.org/56/597656/1/check/openstacksdk-tox-py35-tips/f8d745f/ : SUCCESS in 5m 24s\n- osc-tox-py27-tips http://logs.openstack.org/56/597656/1/check/osc-tox-py27-tips/9a282c7/ : SUCCESS in 7m 55s\n- osc-tox-py35-tips http://logs.openstack.org/56/597656/1/check/osc-tox-py35-tips/3708e43/ : SUCCESS in 7m 33s\n- shade-functional-devstack-tips http://logs.openstack.org/56/597656/1/check/shade-functional-devstack-tips/835380e/ : SUCCESS in 41m 42s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/56/597656/1/check/shade-functional-devstack-tips-python3/18d83f1/ : SUCCESS in 41m 19s\n- shade-tox-py27-tips http://logs.openstack.org/56/597656/1/check/shade-tox-py27-tips/d3e7ae8/ : SUCCESS in 4m 58s\n- shade-tox-py35-tips http://logs.openstack.org/56/597656/1/check/shade-tox-py35-tips/083f69a/ : SUCCESS in 4m 21s\n- openstack-tox-docs http://logs.openstack.org/56/597656/1/check/openstack-tox-docs/5561537/html/ : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/597656/1/check/openstack-tox-lower-constraints/1ac3bb0/ : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"edfd96cd944b313e59476111ef21fcc3591218fc","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-08-30 18:18:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d31468b7defabc2c69ccb43b264013ea20b0779f","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2018-08-30 18:18:46.000000000","message":"Patch Set 1: Code-Review+1\n\nFailure looks like it was caused by an unrelated timeout.","accounts_in_message":[],"_revision_number":1},{"id":"e3eeaef70d0cbe32d341799129806441c8be465f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-30 19:33:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/56/597656/1/check/openstack-tox-pep8/fda4c91/ : SUCCESS in 3m 27s\n- openstack-tox-py27 http://logs.openstack.org/56/597656/1/check/openstack-tox-py27/655a169/ : SUCCESS in 3m 40s\n- openstack-tox-py35 http://logs.openstack.org/56/597656/1/check/openstack-tox-py35/c593034/ : SUCCESS in 3m 39s\n- build-openstack-sphinx-docs http://logs.openstack.org/56/597656/1/check/build-openstack-sphinx-docs/6e0808e/html/ : SUCCESS in 4m 24s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/56/597656/1/check/legacy-tempest-dsvm-neutron-src/1b1c1f5/ : FAILURE in 5m 36s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/56/597656/1/check/openstacksdk-functional-devstack-tips/b92fddd/ : SUCCESS in 1h 08m 51s\n- openstacksdk-functional-devstack-tips-python3 http://logs.openstack.org/56/597656/1/check/openstacksdk-functional-devstack-tips-python3/4591cc2/ : SUCCESS in 1h 09m 00s\n- openstacksdk-tox-py27-tips http://logs.openstack.org/56/597656/1/check/openstacksdk-tox-py27-tips/87e7bf6/ : SUCCESS in 3m 52s\n- openstacksdk-tox-py35-tips http://logs.openstack.org/56/597656/1/check/openstacksdk-tox-py35-tips/e4a1add/ : SUCCESS in 5m 14s\n- osc-tox-py27-tips http://logs.openstack.org/56/597656/1/check/osc-tox-py27-tips/a7c30c2/ : SUCCESS in 5m 24s\n- osc-tox-py35-tips http://logs.openstack.org/56/597656/1/check/osc-tox-py35-tips/bb817bb/ : SUCCESS in 6m 02s\n- shade-functional-devstack-tips http://logs.openstack.org/56/597656/1/check/shade-functional-devstack-tips/2433786/ : SUCCESS in 39m 31s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/56/597656/1/check/shade-functional-devstack-tips-python3/1b9fc2c/ : SUCCESS in 44m 19s\n- shade-tox-py27-tips http://logs.openstack.org/56/597656/1/check/shade-tox-py27-tips/4cf17b3/ : SUCCESS in 3m 33s\n- shade-tox-py35-tips http://logs.openstack.org/56/597656/1/check/shade-tox-py35-tips/8dd9290/ : SUCCESS in 3m 27s\n- openstack-tox-docs http://logs.openstack.org/56/597656/1/check/openstack-tox-docs/a926008/html/ : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/597656/1/check/openstack-tox-lower-constraints/3d39319/ : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c748cf06bacfb489d65531421c3c75848e84ec48","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2018-08-31 07:38:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"598e192f8719d8ccda821a24cfd99b04dcf659cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-08-31 09:17:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/56/597656/1/check/openstack-tox-pep8/c678d53/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/56/597656/1/check/openstack-tox-py27/8199d54/ : SUCCESS in 3m 18s\n- openstack-tox-py35 http://logs.openstack.org/56/597656/1/check/openstack-tox-py35/7fa8918/ : SUCCESS in 3m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/56/597656/1/check/build-openstack-sphinx-docs/ec8c6be/html/ : SUCCESS in 4m 06s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/56/597656/1/check/legacy-tempest-dsvm-neutron-src/f7f58b8/ : SUCCESS in 1h 36m 55s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/56/597656/1/check/openstacksdk-functional-devstack-tips/d9645ae/ : FAILURE in 1h 05m 23s\n- openstacksdk-functional-devstack-tips-python3 http://logs.openstack.org/56/597656/1/check/openstacksdk-functional-devstack-tips-python3/0b5dcd7/ : SUCCESS in 1h 07m 49s\n- openstacksdk-tox-py27-tips http://logs.openstack.org/56/597656/1/check/openstacksdk-tox-py27-tips/c53f884/ : SUCCESS in 3m 26s\n- openstacksdk-tox-py35-tips http://logs.openstack.org/56/597656/1/check/openstacksdk-tox-py35-tips/b6ee3f9/ : SUCCESS in 3m 48s\n- osc-tox-py27-tips http://logs.openstack.org/56/597656/1/check/osc-tox-py27-tips/a4416ce/ : SUCCESS in 5m 47s\n- osc-tox-py35-tips http://logs.openstack.org/56/597656/1/check/osc-tox-py35-tips/136bf02/ : SUCCESS in 7m 17s\n- shade-functional-devstack-tips http://logs.openstack.org/56/597656/1/check/shade-functional-devstack-tips/c0aa8db/ : SUCCESS in 38m 31s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/56/597656/1/check/shade-functional-devstack-tips-python3/f45b836/ : SUCCESS in 37m 23s\n- shade-tox-py27-tips http://logs.openstack.org/56/597656/1/check/shade-tox-py27-tips/cdceb90/ : SUCCESS in 3m 04s\n- shade-tox-py35-tips http://logs.openstack.org/56/597656/1/check/shade-tox-py35-tips/f63bc64/ : SUCCESS in 3m 48s\n- openstack-tox-docs http://logs.openstack.org/56/597656/1/check/openstack-tox-docs/7615842/html/ : SUCCESS in 3m 19s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/597656/1/check/openstack-tox-lower-constraints/d2bdae7/ : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"d9ca5049896603abee9d5a8ca97aacc51d556592","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2018-09-01 02:59:58.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cc51e63cae0c5ae1ac76ccf9b39ddf523cee723f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-01 04:15:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/56/597656/1/check/openstack-tox-pep8/6f55a88/ : SUCCESS in 3m 37s\n- openstack-tox-py27 http://logs.openstack.org/56/597656/1/check/openstack-tox-py27/a998f7b/ : SUCCESS in 3m 35s\n- openstack-tox-py35 http://logs.openstack.org/56/597656/1/check/openstack-tox-py35/8fd3355/ : SUCCESS in 3m 07s\n- build-openstack-sphinx-docs http://logs.openstack.org/56/597656/1/check/build-openstack-sphinx-docs/3b934b1/html/ : SUCCESS in 3m 19s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/56/597656/1/check/legacy-tempest-dsvm-neutron-src/2c3e04c/ : SUCCESS in 1h 13m 54s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/56/597656/1/check/openstacksdk-functional-devstack-tips/4f33238/ : SUCCESS in 56m 21s\n- openstacksdk-functional-devstack-tips-python3 http://logs.openstack.org/56/597656/1/check/openstacksdk-functional-devstack-tips-python3/d747475/ : SUCCESS in 1h 03m 18s\n- openstacksdk-tox-py27-tips http://logs.openstack.org/56/597656/1/check/openstacksdk-tox-py27-tips/e7ad06a/ : SUCCESS in 3m 24s\n- openstacksdk-tox-py35-tips http://logs.openstack.org/56/597656/1/check/openstacksdk-tox-py35-tips/30c1781/ : SUCCESS in 3m 39s\n- osc-tox-py27-tips http://logs.openstack.org/56/597656/1/check/osc-tox-py27-tips/20ede0f/ : SUCCESS in 6m 34s\n- osc-tox-py35-tips http://logs.openstack.org/56/597656/1/check/osc-tox-py35-tips/6168370/ : SUCCESS in 5m 32s\n- shade-functional-devstack-tips http://logs.openstack.org/56/597656/1/check/shade-functional-devstack-tips/b2613c1/ : SUCCESS in 44m 55s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/56/597656/1/check/shade-functional-devstack-tips-python3/91eb39a/ : SUCCESS in 40m 20s\n- shade-tox-py27-tips http://logs.openstack.org/56/597656/1/check/shade-tox-py27-tips/7f28a6e/ : SUCCESS in 3m 11s\n- shade-tox-py35-tips http://logs.openstack.org/56/597656/1/check/shade-tox-py35-tips/b000803/ : SUCCESS in 3m 41s\n- openstack-tox-docs http://logs.openstack.org/56/597656/1/check/openstack-tox-docs/8982a93/html/ : SUCCESS in 2m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/597656/1/check/openstack-tox-lower-constraints/910e325/ : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"e039efdfe2862f6410b9fa34a3aac78525a48fad","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2018-09-03 12:18:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5abddab573f38dc85bf009c2f98501e2b930208f","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2018-09-05 20:14:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73213bb91e6658163d76309a635a322794b664a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-05 20:15:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91bbbf5d399fea4a018b4b78791e21a2bb2ecb47","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-05 21:43:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"db382343bc86b5e3577b7370ed78a20ef9afb052","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-05 21:43:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/56/597656/1/gate/openstack-tox-pep8/141ab52/ : SUCCESS in 3m 28s\n- openstack-tox-py27 http://logs.openstack.org/56/597656/1/gate/openstack-tox-py27/8a7fa74/ : SUCCESS in 3m 28s\n- openstack-tox-py35 http://logs.openstack.org/56/597656/1/gate/openstack-tox-py35/675174f/ : SUCCESS in 4m 17s\n- build-openstack-sphinx-docs http://logs.openstack.org/56/597656/1/gate/build-openstack-sphinx-docs/7492b35/html/ : SUCCESS in 3m 13s\n- legacy-tempest-dsvm-neutron-src http://logs.openstack.org/56/597656/1/gate/legacy-tempest-dsvm-neutron-src/f90fb31/ : SUCCESS in 1h 21m 17s\n- openstacksdk-functional-devstack-tips http://logs.openstack.org/56/597656/1/gate/openstacksdk-functional-devstack-tips/5a90d41/ : SUCCESS in 1h 12m 30s\n- openstacksdk-functional-devstack-tips-python3 http://logs.openstack.org/56/597656/1/gate/openstacksdk-functional-devstack-tips-python3/c933fee/ : SUCCESS in 1h 08m 55s\n- openstacksdk-tox-py27-tips http://logs.openstack.org/56/597656/1/gate/openstacksdk-tox-py27-tips/8b3b17f/ : SUCCESS in 4m 07s\n- openstacksdk-tox-py35-tips http://logs.openstack.org/56/597656/1/gate/openstacksdk-tox-py35-tips/70b1527/ : SUCCESS in 3m 45s\n- osc-tox-py27-tips http://logs.openstack.org/56/597656/1/gate/osc-tox-py27-tips/e42a43f/ : SUCCESS in 6m 34s\n- osc-tox-py35-tips http://logs.openstack.org/56/597656/1/gate/osc-tox-py35-tips/53f5bb2/ : SUCCESS in 5m 15s\n- shade-functional-devstack-tips http://logs.openstack.org/56/597656/1/gate/shade-functional-devstack-tips/6b174eb/ : SUCCESS in 48m 25s\n- shade-functional-devstack-tips-python3 http://logs.openstack.org/56/597656/1/gate/shade-functional-devstack-tips-python3/198a5a6/ : SUCCESS in 46m 05s\n- shade-tox-py27-tips http://logs.openstack.org/56/597656/1/gate/shade-tox-py27-tips/752c3bc/ : SUCCESS in 3m 50s\n- shade-tox-py35-tips http://logs.openstack.org/56/597656/1/gate/shade-tox-py35-tips/df4c953/ : SUCCESS in 4m 45s\n- openstack-tox-docs http://logs.openstack.org/56/597656/1/gate/openstack-tox-docs/a7c23a5/html/ : SUCCESS in 3m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/56/597656/1/gate/openstack-tox-lower-constraints/ede6042/ : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d829d3d69b1920fbad6972c9a70c771f68e43c3d","revisions":{"d829d3d69b1920fbad6972c9a70c771f68e43c3d":{"kind":"REWORK","_number":1,"created":"2018-08-29 20:10:53.000000000","uploader":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"ref":"refs/changes/56/597656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/56/597656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/56/597656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/56/597656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/56/597656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/56/597656/1"}}},"commit":{"parents":[{"commit":"cf83ba0ae314ef3b76682ca468f8e7980be5f466","subject":"import zuul job settings from project-config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/cf83ba0ae314ef3b76682ca468f8e7980be5f466"}]}],"author":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-08-29 20:04:48.000000000","tz":-240},"committer":{"name":"Doug Hellmann","email":"doug@doughellmann.com","date":"2018-08-29 20:04:48.000000000","tz":-240},"subject":"switch documentation job to new PTI","message":"switch documentation job to new PTI\n\nThis is a mechanically generated patch to switch the documentation\njobs to use the new PTI versions of the jobs as part of the\npython3-first goal.\n\nSee the python3-first goal document for details:\nhttps://governance.openstack.org/tc/goals/stein/python3-first.html\n\nChange-Id: I30f3bc0f5b68190f72b015ae5a14251c91eeee42\nStory: #2002586\nTask: #24304\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/d829d3d69b1920fbad6972c9a70c771f68e43c3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/d829d3d69b1920fbad6972c9a70c771f68e43c3d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"cf83ba0ae314ef3b76682ca468f8e7980be5f466","is_merged_in_target_branch":true,"change_id":"Iab3d02dbf7b6df016bf44f7be7b27c01f498601e","change_number":597655,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
