)]}'
{"id":"openstack%2Fkeystoneauth~681103","triplet_id":"openstack%2Fkeystoneauth~master~I2a28f3f4eaabaa772df395f3f5d55b6fd78f8968","project":"openstack/keystoneauth","branch":"master","topic":"ftbfs2020","hashtags":[],"change_id":"I2a28f3f4eaabaa772df395f3f5d55b6fd78f8968","subject":"Make tests pass in 2020","status":"MERGED","created":"2019-09-09 19:48:45.000000000","updated":"2020-01-07 18:33:51.000000000","submitted":"2019-10-05 09:51:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"681103-1570269107323-adebb769","meta_rev_id":"a87ce5f3b4b26d1dd725a2d7543979cc6782743b","_number":681103,"virtual_id_number":681103,"owner":{"_account_id":1446,"name":"Bernhard M.","email":"bmwiedemann@zq1.de","username":"ubuntubmw","status":"dev at SUSE"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"date":"2019-09-27 12:24:17.000000000","_account_id":1446,"name":"Bernhard M.","email":"bmwiedemann@zq1.de","username":"ubuntubmw","status":"dev at SUSE"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-05 09:51:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-09-27 18:10:51.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-10-05 07:57:34.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":1446,"name":"Bernhard M.","email":"bmwiedemann@zq1.de","username":"ubuntubmw","status":"dev at SUSE"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2019-10-05 07:57:34.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":1446,"name":"Bernhard M.","email":"bmwiedemann@zq1.de","username":"ubuntubmw","status":"dev at SUSE"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":1446,"name":"Bernhard M.","email":"bmwiedemann@zq1.de","username":"ubuntubmw","status":"dev at SUSE"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-11 21:18:35.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REVIEWER"},{"updated":"2019-09-27 11:27:28.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-09-27 18:10:51.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-10-05 07:57:34.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2019-10-05 09:51:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6b7d8a510f3ca015c803a163ec5b9caedb432fc9","author":{"_account_id":1446,"name":"Bernhard M.","email":"bmwiedemann@zq1.de","username":"ubuntubmw","status":"dev at SUSE"},"date":"2019-09-09 19:48:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ceedea71643c9fb3c3fbcc445bd00fa59025cdb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 21:49:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/81113f8116f54faab72c0a7ad7cf5971 : SUCCESS in 1h 55m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/11f93b3c2b8840ac8f4a340fd19a76b7 : SUCCESS in 1h 30m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/849761ebd0b447e383168ab7f4c710c9 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bde7e61874e41a784a6d19086227c07 : SUCCESS in 2m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a7922339f80741fa939154570d4d022d : SUCCESS in 3m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dec73a69de4044f7b382a1d1b781e1d8 : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d63b1d0fa9b549c29e7cf9fc5a063f1c : SUCCESS in 4m 57s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f6b3ef3c61f64704805af320675a2150 : SUCCESS in 55m 59s\n- openstacksdk-functional-devstack-tips-python2 https://zuul.opendev.org/t/openstack/build/423c838739cd485d8ded8f0de8bfa964 : SUCCESS in 1h 02m 30s\n- openstacksdk-tox-py36-tips https://zuul.opendev.org/t/openstack/build/828406b67d684dc4aadcfae90976d9cb : SUCCESS in 4m 19s\n- osc-tox-py27-tips https://zuul.opendev.org/t/openstack/build/8c73219efbac4b9fae7cbbee7625aa6b : SUCCESS in 6m 42s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/53a9a527911c49f29003affcdca85137 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4ef4d2a109a4252ba6d319d7c539972 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d69f59bc3d1eb784a767e2dca2e079eb7f5e162f","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-09-11 21:18:35.000000000","message":"Patch Set 1: Code-Review+2\n\nHopefully someone remembers to update this again in 980 years.","accounts_in_message":[],"_revision_number":1},{"id":"4ebfa231390c4af61a6269662931d51ef38bd42f","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-16 22:02:43.000000000","message":"Patch Set 1: Code-Review-1\n\nMorgan had feedback on a similar patch https://review.opendev.org/657780 that was never addressed, could you address that first and then ensure this patch is consistent with the other one?","accounts_in_message":[],"_revision_number":1},{"id":"87fb9459b6c57f694ea73177a7cb82ad46e05a7f","author":{"_account_id":1446,"name":"Bernhard M.","email":"bmwiedemann@zq1.de","username":"ubuntubmw","status":"dev at SUSE"},"date":"2019-09-17 08:43:11.000000000","message":"Patch Set 1:\n\nupdated 657780 - want to review that before I spend time doing the same here?","accounts_in_message":[],"_revision_number":1},{"id":"d8bef338b391d5407076496f94b1399742e6585c","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-17 14:17:16.000000000","message":"Patch Set 1:\n\n\u003e updated 657780 - want to review that before I spend time doing the\n \u003e same here?\n\n+2","accounts_in_message":[],"_revision_number":1},{"id":"f17f99fd6a5018808922c286d0b82adc93bab8e6","author":{"_account_id":1446,"name":"Bernhard M.","email":"bmwiedemann@zq1.de","username":"ubuntubmw","status":"dev at SUSE"},"date":"2019-09-27 08:45:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"55543844eea9d3b3b47f9f667df19aa9ed9e052a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 10:15:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c379774d4ff345859ed17fd81b152f14 : FAILURE in 1h 23m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc5118d216e744f7af05ae6abe0fb111 : SUCCESS in 1h 23m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/602b2f77f9de4d5f924e7b3decb8ebac : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ca73ac98d3947f89257a861026d828b : FAILURE in 3m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/58b5407519924ce381fbd7cb5d61f35a : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d79ac3f5c7f4bd591d4d83719a629f4 : SUCCESS in 3m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b878c8999ca342f2a008aa70f9e9ac36 : SUCCESS in 5m 01s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/55691eb904234e35b0de721bd7e44a58 : SUCCESS in 1h 03m 28s\n- openstacksdk-functional-devstack-tips-python2 https://zuul.opendev.org/t/openstack/build/3ac8193775a24a47adb29f226c5f21aa : SUCCESS in 1h 02m 18s\n- openstacksdk-tox-py36-tips https://zuul.opendev.org/t/openstack/build/6ade8505cc424e488b7611392994ce8b : SUCCESS in 3m 33s\n- osc-tox-py27-tips https://zuul.opendev.org/t/openstack/build/7b96afc165464bcdaaa29fdb8b1f96be : SUCCESS in 4m 18s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/ead63c246d5140948f90c859bbcf8130 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4af4bc0ba2e84e46b77f68968a6aaa18 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"f1033bd12252782b0b91aebef8cc8149f8b6a8fb","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-27 11:27:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7f96c03e57d651267548892a948f9c94bdcd98fc","author":{"_account_id":1446,"name":"Bernhard M.","email":"bmwiedemann@zq1.de","username":"ubuntubmw","status":"dev at SUSE"},"date":"2019-09-27 11:59:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bc2e3c01d5cdf9a00c88e1ade12a963042e53aaa","author":{"_account_id":1446,"name":"Bernhard M.","email":"bmwiedemann@zq1.de","username":"ubuntubmw","status":"dev at SUSE"},"date":"2019-09-27 12:24:17.000000000","message":"Patch Set 3:\n\n@Vishakha : it seems that line does not cause test failures, but I patched it anyway for consistency.\nThere are 4 more occurences of 2020-01-01 in other files, that seem to not matter. Leave or update?","accounts_in_message":[],"_revision_number":3},{"id":"1758a1f4e5228b8e58997dee3486621d07199424","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 13:59:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/0f43d9d5281f4f65b846776daad0994c : SUCCESS in 1h 57m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6b9c9c959a64e1487c24047cfe54674 : SUCCESS in 1h 28m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47b02fa3e5ae48e2889052bf96ae82c4 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/657ef2585e954340b0eb3f5bfbd2ad2b : SUCCESS in 2m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/004ebf6eb0574fd78c7960c0a1703255 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3ea3ffab5b34171be3c577f3299e03e : SUCCESS in 3m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a76177b31bf44a49725b9ba16564bed : SUCCESS in 3m 25s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f60b6869ba7446fdb7dbdf90ba27e92d : SUCCESS in 1h 00m 40s\n- openstacksdk-functional-devstack-tips-python2 https://zuul.opendev.org/t/openstack/build/799500d73d2e4fef81c14aeab73b3c43 : SUCCESS in 58m 03s\n- openstacksdk-tox-py36-tips https://zuul.opendev.org/t/openstack/build/0b11c5cb67f2405aadf1dbf5a8a5c8e6 : FAILURE in 4m 02s\n- osc-tox-py27-tips https://zuul.opendev.org/t/openstack/build/e95836e98f8d4198a73d2176542c0cb8 : SUCCESS in 4m 14s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/dbeb30a7fddd4762a69a7cf9a2991c11 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47bd9f9f9f1e4283b60bbf9b0f8a106e : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":3},{"id":"97a5631babc0c9d5790e109262dac2344d33c604","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-09-27 18:10:51.000000000","message":"Patch Set 3: Code-Review+2\n\nrecheck\n\n \u003e @Vishakha : it seems that line does not cause test failures, but I\n \u003e patched it anyway for consistency.\n \u003e There are 4 more occurences of 2020-01-01 in other files, that seem\n \u003e to not matter. Leave or update?\n\nIf they don\u0027t actually test authentication and then the date string does not really matter so I would not bother to change them.","accounts_in_message":[],"_revision_number":3},{"id":"96ca2465cd9a22e94e63047d327d2529064b13fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 20:21:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/525b0daa8ff2420d90dad061077ac60e : SUCCESS in 2h 03m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88b66d618a1e47d4b78aedf1fa896a38 : SUCCESS in 1h 18m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15aefa69bfa24e018b773d3343313e7a : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55dfba4f3c714e94a6ef6f8e36a3b083 : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/168110ff5c0f4dd681df5805613c3207 : SUCCESS in 3m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a1bc2ee563d4c5ab498cfcb7a32c9a7 : SUCCESS in 3m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e33e27028ea4c9db0d233c395ab7e8c : SUCCESS in 3m 31s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/fb36c7c0e19f4324b66830dd31cee9b3 : SUCCESS in 1h 04m 43s\n- openstacksdk-functional-devstack-tips-python2 https://zuul.opendev.org/t/openstack/build/2708db1630ce4beb9a9ab934ce67d717 : SUCCESS in 1h 09m 37s\n- openstacksdk-tox-py36-tips https://zuul.opendev.org/t/openstack/build/f5c193a9cce0470f9422f22dfe024e2a : SUCCESS in 4m 05s\n- osc-tox-py27-tips https://zuul.opendev.org/t/openstack/build/4e7a942b4a3b4a77a0f2c18fbb25350e : SUCCESS in 4m 00s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/b327fc1596e64c188e7b40159fd01db2 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab5b1cd0f1ab4ced9589c3fa65a6d628 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":3},{"id":"7c27b420c6ebd50fa207d07fe3955236ecd71428","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-10-05 07:57:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"562129e700134f3f8d7e25b833be00c179e286fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 07:57:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"48259c967056e521a857468c46b3b9502690ab14","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 09:51:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d2c6d1fe282aa39713f953ee1a00818e212a6692","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 09:51:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/1d70f90044ba419488c404af63071a8e : SUCCESS in 1h 46m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8906553519964582a8cc7b3b2f4ba567 : SUCCESS in 1h 25m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20c13246dba241438098833e00f7bd08 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef4f5256edcd4c7988968478d56e17a9 : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f94535e7c594425b9514ea9ea86d063 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fbaa76e96b3492284a06986c32f9c20 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65f344b6f00245afba0c5ca87fe2b2a5 : SUCCESS in 4m 05s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c69e92a7442444cc803a09f7ab451530 : SUCCESS in 1h 08m 52s\n- openstacksdk-functional-devstack-tips-python2 https://zuul.opendev.org/t/openstack/build/69af322dbd8f44c093a82bbc04fefc20 : SUCCESS in 1h 11m 01s\n- openstacksdk-tox-py36-tips https://zuul.opendev.org/t/openstack/build/9bb0c5d6f2204c85897d89351da2637d : SUCCESS in 5m 27s\n- osc-tox-py27-tips https://zuul.opendev.org/t/openstack/build/22c9eea5c5554be1bd25e390757d4206 : SUCCESS in 4m 56s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/a9a218b96aab43a6bceade81b48fddaa : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/debef84739434b53813fcfbb03dc3c54 : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":3},{"id":"c8de740adc405eadf0dd70a9163cc945404e6fc6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 09:53:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/396ff5193d5a439394df2f220e40cad0 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":3},{"id":"3e72c95b515e2313269707b2d2aba749101fa56b","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-01-07 18:33:51.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 00eda6b19e34e896b7825c9d017c79b1b465dd9a","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4461358098cbbe3ce27f05801ae06b191b092d05","revisions":{"2ada90821c3decfbfc39bfe32f4cc8a40e845108":{"kind":"REWORK","_number":1,"created":"2019-09-09 19:48:45.000000000","uploader":{"_account_id":1446,"name":"Bernhard M.","email":"bmwiedemann@zq1.de","username":"ubuntubmw","status":"dev at SUSE"},"ref":"refs/changes/03/681103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/03/681103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/03/681103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/03/681103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/03/681103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/03/681103/1"}}},"commit":{"parents":[{"commit":"8b06c57292753e920aac9407a2fa9659adb5c307","subject":"Simplify session logger object tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/8b06c57292753e920aac9407a2fa9659adb5c307"}]}],"author":{"name":"Bernhard M. Wiedemann","email":"bwiedemann@suse.de","date":"2019-09-09 19:00:14.000000000","tz":120},"committer":{"name":"Bernhard M. Wiedemann","email":"bwiedemann@suse.de","date":"2019-09-09 19:00:14.000000000","tz":120},"subject":"Make tests pass in 2020","message":"Make tests pass in 2020\n\nThis issue was found while working on reproducible builds for openSUSE.\n\nChange-Id: I2a28f3f4eaabaa772df395f3f5d55b6fd78f8968\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/2ada90821c3decfbfc39bfe32f4cc8a40e845108"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/2ada90821c3decfbfc39bfe32f4cc8a40e845108"}]},"branch":"refs/heads/master"},"c2d27a77117ea53aaa5b47e36ad6f1b1fd348053":{"kind":"REWORK","_number":2,"created":"2019-09-27 08:45:21.000000000","uploader":{"_account_id":1446,"name":"Bernhard M.","email":"bmwiedemann@zq1.de","username":"ubuntubmw","status":"dev at SUSE"},"ref":"refs/changes/03/681103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/03/681103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/03/681103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/03/681103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/03/681103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/03/681103/2"}}},"commit":{"parents":[{"commit":"b59ef48a35028fe0804907612e73cf792e9109d2","subject":"Fix misspell word","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/b59ef48a35028fe0804907612e73cf792e9109d2"}]}],"author":{"name":"Bernhard M. Wiedemann","email":"bwiedemann@suse.de","date":"2019-09-09 19:00:14.000000000","tz":120},"committer":{"name":"Bernhard M. Wiedemann","email":"bwiedemann@suse.de","date":"2019-09-27 08:44:37.000000000","tz":120},"subject":"Make tests pass in 2020","message":"Make tests pass in 2020\n\nThis issue was found while working on reproducible builds for openSUSE.\n\nThis solves it similar to change I73bde68be53afff4e8dff12d756b8381f34b2adb\n\nChanged month to February to avoid races around new year.\n\nChange-Id: I2a28f3f4eaabaa772df395f3f5d55b6fd78f8968\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/c2d27a77117ea53aaa5b47e36ad6f1b1fd348053"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/c2d27a77117ea53aaa5b47e36ad6f1b1fd348053"}]},"branch":"refs/heads/master"},"4461358098cbbe3ce27f05801ae06b191b092d05":{"kind":"REWORK","_number":3,"created":"2019-09-27 11:59:12.000000000","uploader":{"_account_id":1446,"name":"Bernhard M.","email":"bmwiedemann@zq1.de","username":"ubuntubmw","status":"dev at SUSE"},"ref":"refs/changes/03/681103/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/03/681103/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/03/681103/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/03/681103/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/03/681103/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/03/681103/3"}}},"commit":{"parents":[{"commit":"b59ef48a35028fe0804907612e73cf792e9109d2","subject":"Fix misspell word","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/b59ef48a35028fe0804907612e73cf792e9109d2"}]}],"author":{"name":"Bernhard M. Wiedemann","email":"bwiedemann@suse.de","date":"2019-09-09 19:00:14.000000000","tz":120},"committer":{"name":"Bernhard M. Wiedemann","email":"bwiedemann@suse.de","date":"2019-09-27 11:58:34.000000000","tz":120},"subject":"Make tests pass in 2020","message":"Make tests pass in 2020\n\nThis issue was found while working on reproducible builds for openSUSE.\n\nThis solves it similar to change I73bde68be53afff4e8dff12d756b8381f34b2adb\n\nChanged month to February to avoid races around new year.\n\nChange-Id: I2a28f3f4eaabaa772df395f3f5d55b6fd78f8968\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/4461358098cbbe3ce27f05801ae06b191b092d05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/4461358098cbbe3ce27f05801ae06b191b092d05"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
