)]}'
{"id":"openstack%2Fkeystoneauth~743227","triplet_id":"openstack%2Fkeystoneauth~master~Iead4b95c1f5b8d84cec705da32f41049e2eea641","project":"openstack/keystoneauth","branch":"master","hashtags":[],"change_id":"Iead4b95c1f5b8d84cec705da32f41049e2eea641","subject":"Fix get_endpoint_data for non-keystone plugins","status":"MERGED","created":"2020-07-27 13:26:50.000000000","updated":"2020-07-28 14:25:32.000000000","submitted":"2020-07-28 14:22:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743227-1595946167418-8e31e638","meta_rev_id":"d1b49189475c3c693e7c12748405e9b284ca422a","_number":743227,"virtual_id_number":743227,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-28 14:22:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-28 10:51:24.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-07-28 12:39:30.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-07-28 12:39:30.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-28 10:51:24.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-07-28 14:22:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4295cf096f78e3f88fbdaab4f481388c11d49dbf","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-07-27 13:26:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64b793a0f77f710b03489c01effcc11b6ca00db9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 14:27:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6c0b777cf8c44c7bfa999f5f71db1cc : SUCCESS in 55m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82ef059348dc45dda8071fc5a9c458cf : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d997988ef4f448bfae0e08374b60ba90 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8fd508353b84000bbc41660455555dc : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e36b881bedb4bd885d4eae9e9b6a3b0 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31c716a32bad4501a258d2476b0f7f4a : SUCCESS in 6m 23s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/256f6e8daf72459384551e5b34d562c1 : SUCCESS in 38m 09s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/3fb92c1b182d4c2baa33d42d18a7c6c5 : SUCCESS in 6m 02s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/5b8f5164ce1a44d58bdb9f4b07eaa557 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd1ea4cafa242ba8f367a7c94a3e4e8 : FAILURE in 7m 23s","accounts_in_message":[],"_revision_number":1},{"id":"8d7a8fa45a3705a8fb4f982ccd023b0f269eeed1","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-27 15:11:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f991fd688b7975186db2f4f243c8a94ceb1a166c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-27 15:21:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4ce610b36bd055fd67dd4a0e6f3a583b049e3e13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 16:50:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5539034674e44ce8edd4158b4eb02f2 : SUCCESS in 1h 19m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5ad6530376946938c97bc6f7221c244 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1ce2b2ae34b4dd1afcb389b7d291841 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd269390628b4592bc389aaf73f3f769 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59685a9fd8de4b6fbebcedb80190470c : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96cb8114f40c4e4db4e228f8d5add11c : SUCCESS in 4m 59s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/4cb0ebab79944ea2932ffb30ae1bcdd5 : SUCCESS in 41m 36s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/66c7cdd873294831a490c53c354ee2d3 : SUCCESS in 5m 31s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/c38a5fe752a3434d98de70cc69bac981 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aaa6c49adff4969b4cc00a18d6bf9bd : FAILURE in 5m 16s","accounts_in_message":[],"_revision_number":2},{"id":"4728e86e8695298f6bf71f6726a89ff1b0ea4671","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-28 10:51:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ca93ed9bbcc1d3224bd3aa6ef0b687530daf7d60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 12:30:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1bc933c3c164c14b8032ae97ad9d5df : SUCCESS in 1h 31m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cd7b95fe3a34a8193b946c212ec64a5 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/418776819d844b578e87322b1fd45c58 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fb45703f00641af9a491a5a2f3f8f81 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/642135ff55ff4d1aaff7ee6775338136 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0c8500b5b7147f2a8ec2294fe1b6f14 : SUCCESS in 5m 33s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7092a97748b44d64a72f09d0351e9c3e : SUCCESS in 56m 42s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/b9526fa51fd9434280bb20664b9241c9 : SUCCESS in 6m 04s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/9c43e9a7f4354ef78d2e8969e611324a : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6908f0edb62b4adf8e282bee094a6550 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"062805040c9f5ec28cebe34f6f7121b62f5d5288","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-07-28 12:39:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b6507c6993aa22c523af53b64c7cfe61de18fb5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 12:40:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6d5e14e8177e55b9b355eec031b1cdfab9879d51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 14:22:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac4a6097a01f4835b1750663d23e3580 : SUCCESS in 1h 38m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd2d29b7b1154944896145d60f72f3d7 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72e0a70b8c41430e9e417c59c1c52969 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd61044191864864bd7d455481d03e86 : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e40ee5139d1495589574a4254708534 : SUCCESS in 4m 51s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9d46c89598bb4ca5b6216802a87c1b2e : SUCCESS in 1h 02m 56s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/a800808108574d3c969e1e923daaba13 : SUCCESS in 5m 41s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/8d75540dc8a5420cb79cc2204d35b3ae : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31902c8f0e77407d99a1554f2986ae69 : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":2},{"id":"b41c6e8f7bda82c3bd14e3fa0d444454254ebfb4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 14:22:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d1b49189475c3c693e7c12748405e9b284ca422a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 14:25:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce95b7aba40a4710b9476441f3b07d34 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b95a89e3ff65f175b0b4b422ffa968d676adacbf","revisions":{"ab9021ce929ef505a12187abbc1e1c9f168acf16":{"kind":"REWORK","_number":1,"created":"2020-07-27 13:26:50.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/27/743227/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/27/743227/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/27/743227/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/27/743227/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/27/743227/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/27/743227/1"}}},"commit":{"parents":[{"commit":"dab8e1057ae8bb9a0e778fb8d3141ad4fb36a339","subject":"Merge \"Drop python 3.5 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/dab8e1057ae8bb9a0e778fb8d3141ad4fb36a339"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-07-27 13:25:41.000000000","tz":-300},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-07-27 13:25:41.000000000","tz":-300},"subject":"Add unit test for httpbasicauth and discovery","message":"Add unit test for httpbasicauth and discovery\n\nIn bifrost we\u0027re seeing issues with get_api_major_version and\nhttpbasic.\n\nChange-Id: Iead4b95c1f5b8d84cec705da32f41049e2eea641\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/ab9021ce929ef505a12187abbc1e1c9f168acf16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/ab9021ce929ef505a12187abbc1e1c9f168acf16"}]},"branch":"refs/heads/master"},"b95a89e3ff65f175b0b4b422ffa968d676adacbf":{"kind":"REWORK","_number":2,"created":"2020-07-27 15:21:07.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/27/743227/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/27/743227/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/27/743227/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/27/743227/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/27/743227/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/27/743227/2"}}},"commit":{"parents":[{"commit":"dab8e1057ae8bb9a0e778fb8d3141ad4fb36a339","subject":"Merge \"Drop python 3.5 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/dab8e1057ae8bb9a0e778fb8d3141ad4fb36a339"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2020-07-27 13:25:41.000000000","tz":-300},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-27 15:20:58.000000000","tz":120},"subject":"Fix get_endpoint_data for non-keystone plugins","message":"Fix get_endpoint_data for non-keystone plugins\n\nWe expect endpoint_override, but these plugins won\u0027t necessary\nhave it, they have endpoint instead.\n\nCo-Authored-By: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nChange-Id: Iead4b95c1f5b8d84cec705da32f41049e2eea641\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/b95a89e3ff65f175b0b4b422ffa968d676adacbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/b95a89e3ff65f175b0b4b422ffa968d676adacbf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
