)]}'
{"id":"openstack%2Fkeystoneauth~743282","triplet_id":"openstack%2Fkeystoneauth~master~Ie623931b150748d7759cf276e0023a2f06a8d4db","project":"openstack/keystoneauth","branch":"master","topic":"noauth-discovery","hashtags":[],"change_id":"Ie623931b150748d7759cf276e0023a2f06a8d4db","subject":"Correct major version discovery for non-keystone plugins","status":"MERGED","created":"2020-07-27 16:59:31.000000000","updated":"2020-07-31 18:01:12.000000000","submitted":"2020-07-31 17:59:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743282-1596218390668-0abe18e9","meta_rev_id":"e402c0fefa85162b44011690f0564f38e376b85a","_number":743282,"virtual_id_number":743282,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-31 17:59:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2020-07-31 14:11:14.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-07-31 16:21:09.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2020-07-31 16:21:09.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-29 07:58:53.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-07-31 14:11:14.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-07-31 16:21:09.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-07-31 17:59:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"322dde266bf26881b1817283a59b4f056356a688","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-27 16:59:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb1db110962c4c649254ab223f3fdf0f90b68c56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 18:16:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65bcfe38e3db43049fa058f4a3ab1116 : FAILURE in 1h 10m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/871c47eed1eb4868b35e1d6f645860b5 : FAILURE in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f406bf4fcb49470eabd7d4c4f8117e78 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46c8fcc4dd2b441bb98d411b7e6efd18 : FAILURE in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/842988160e084dbfa65a8c0e809a5bad : FAILURE in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b6b75ced295437f84fcae6d867e9ea2 : FAILURE in 5m 02s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0489a5faecdf404581b962e47ecd9a60 : SUCCESS in 1h 00m 21s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/1b3e3fe91939422499ab88635aab26f7 : SUCCESS in 5m 43s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/27de9ce036624de6be1a6360043cb2b7 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/233b4f2939c54eeb8c9c6e84db276039 : FAILURE in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"fd9113e0739535bea43db2aac3100a9820e356d6","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-28 10:55:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"72ceb6a2caaed3d4433cd197a61ae62cfbe6f6af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 12:17:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/912bfb6c80574bffa51a193d51943293 : SUCCESS in 1h 18m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b722e6fb92294e97ad14f613b1be1dcc : FAILURE in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84c7ea3cc13a4628ace047bab75ca41b : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48948541ec634b388c8e367e83c33e0c : FAILURE in 3m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2d889dbf4a2943459b038b734a3a539a : FAILURE in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27b99102b64345958ae111f48d69631d : FAILURE in 4m 17s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/90a71f4869ee4028bbf08144e5d0426e : SUCCESS in 49m 54s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/c27502c26c344d16bb5eb218400142ee : SUCCESS in 5m 00s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/505cefa91b3047e38b214e0af34b8d39 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/420567d3ba1d4f3a844501abea9cb987 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"0df5418f02f8329653b0d0b3d6e258d6e08b7bf4","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-28 13:00:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f632e60071d0bbd5d15499bad03edfce65e06797","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 14:28:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00c7e33a289944efa0daee413f704261 : SUCCESS in 1h 20m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b168d8a5fb648ef8df21496b71816a3 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a23e438834dd42e4af9fa22efdb455b9 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/478c5908f28c40aa9ce0dd0296736fbf : SUCCESS in 3m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8814a845249c484084d03544c87b784f : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e42ac64c09aa4b5c9a053f85813b2473 : SUCCESS in 5m 11s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f3e0ec60e69344dfaed2cbc7a791ef34 : SUCCESS in 41m 02s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/dd8fbc2883de4a849c1bd6ff0215e16b : SUCCESS in 4m 57s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/056b7a9359a945be8d014461fb49eafe : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d01c157a9e5642848bdc2cab635a6063 : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":2},{"id":"d1a8f46da64809ea3e7ac0016c9eef1716e3eb6e","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-28 14:50:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bd48d1c2d233e8fe85350b8d5acc5b1c213323b7","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-07-28 14:52:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b068298dc3fca193fe94bfab6139e473f454244b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-28 16:17:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef8e3b5bc234442f9dfadefe628ea652 : SUCCESS in 1h 14m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c0d9aad92ca4c65a403d628ce769037 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/164a8b6dc16b405697191b1a6ec037b5 : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ebbb68ac506449aae168ecfd70ac532 : SUCCESS in 4m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f2fbd3524f324690b625615ca1dea3c0 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46c3d77954ee4b2aa65703f2298e2073 : SUCCESS in 4m 46s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/14a083afa92b446ca83d891256b220af : SUCCESS in 56m 02s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/150818c9e66944dfb313772b11a0320b : SUCCESS in 7m 35s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/a97ae7a90a904949b578ebcd315715b5 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fd4f19c130a4b90868a9c5ac124e88d : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6977ee17db7545739b87e1b197b50ed5 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":3},{"id":"905afa6470dd9c48b1dda7e9e9f9b8a1512259ae","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-07-30 20:42:44.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"73c4abbee56209f9a48011a5080c34933995cf41","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-31 09:29:28.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"23bac382b7034b171dcc6bb192d2cc382c2da017","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-31 09:32:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7f867f411585453f5b41be5e62e54df33816d281","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 10:59:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37de5a181a3d44088c9e7daba391fb47 : SUCCESS in 1h 20m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c488c050d50840e0b23769a75f0fd83e : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f86708613dee487b99771d407328e9df : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9481b30049f243cea666d7ec344160bd : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/91dbc37a64ae487ba9d289829e93da59 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd972dc729024ae98d90178029e0881b : SUCCESS in 5m 02s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/20da42e8198e4081af9e49d55c9ef385 : SUCCESS in 41m 39s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/757edd3fb0bc41de8b2cf5eca80255c9 : SUCCESS in 5m 25s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/4d7a98a48b374ff1ba47816f9a6c721e : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5916add875224b63983f292e972e2024 : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57ec329944a044d0a45dfd95e7ee1fd9 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":4},{"id":"b39e35a5aafea4706ff78e334951b5845eb63724","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-07-31 14:11:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"358144e885f7842ca1164d85cd11999354df6419","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2020-07-31 16:21:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d0db2323bf99c05f49f9e7d76d0a29821ba4ad9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 16:21:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"05dc02b91c82ce5bbd2bf96150262c01a6367be7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 17:59:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c48611b11f363c1e34d01490502003e04cc73935","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 17:59:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b6fdbe853ed46efa822aa0a0035676b : SUCCESS in 1h 31m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08618a7ae37a4eac9ef01fb3ef557e6f : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11e7e2005d9440e0901e800779717ab5 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4c0ac8025b9471788df634dfa5cb175 : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d32d18c141a341118c060d4c098c3a17 : SUCCESS in 4m 38s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c7fd8742f753433e9d913d4a4d622129 : SUCCESS in 56m 06s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/063d97caed384033a3d3d44431b1531b : SUCCESS in 4m 43s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/7c7bb6de084442c883ade95d3ea449a8 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c52205458eb64412a63092e75920c788 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d200ba4a0904a329a56417c0aeeeea8 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":4},{"id":"6cd292312cf66ded1b8a8898085a1ae5423acd97","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 18:01:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4948c6a7592f40f1866b780b37f63fee : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f9f064c9bbc48e2b3250dfd8d97b604 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"981a19bba1d5a8477d51ed4751c87a3a4d130aee","revisions":{"7fb3f54a450e02685c3f00de22d51123cbba5fdc":{"kind":"REWORK","_number":1,"created":"2020-07-27 16:59:31.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/82/743282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/82/743282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/82/743282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/82/743282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/82/743282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/82/743282/1"}}},"commit":{"parents":[{"commit":"b95a89e3ff65f175b0b4b422ffa968d676adacbf","subject":"Fix get_endpoint_data for non-keystone plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/b95a89e3ff65f175b0b4b422ffa968d676adacbf"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-27 16:58:44.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-27 16:58:44.000000000","tz":120},"subject":"[WIP] Correct endpoint discovery for non-keystone plugins","message":"[WIP] Correct endpoint discovery for non-keystone plugins\n\nChange-Id: Ie623931b150748d7759cf276e0023a2f06a8d4db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/7fb3f54a450e02685c3f00de22d51123cbba5fdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/7fb3f54a450e02685c3f00de22d51123cbba5fdc"}]},"branch":"refs/heads/master"},"e0f381950e455c74fec6a58f339dac65f0fd4005":{"kind":"REWORK","_number":2,"created":"2020-07-28 13:00:41.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/82/743282/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/82/743282/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/82/743282/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/82/743282/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/82/743282/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/82/743282/2"}}},"commit":{"parents":[{"commit":"b95a89e3ff65f175b0b4b422ffa968d676adacbf","subject":"Fix get_endpoint_data for non-keystone plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/b95a89e3ff65f175b0b4b422ffa968d676adacbf"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-27 16:58:44.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-28 13:00:26.000000000","tz":120},"subject":"[WIP] Correct endpoint discovery for non-keystone plugins","message":"[WIP] Correct endpoint discovery for non-keystone plugins\n\nChange-Id: Ie623931b150748d7759cf276e0023a2f06a8d4db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/e0f381950e455c74fec6a58f339dac65f0fd4005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/e0f381950e455c74fec6a58f339dac65f0fd4005"}]},"branch":"refs/heads/master"},"5df761193e8333f0029627d23ac88904c5c28b08":{"kind":"REWORK","_number":3,"created":"2020-07-28 14:50:17.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/82/743282/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/82/743282/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/82/743282/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/82/743282/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/82/743282/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/82/743282/3"}}},"commit":{"parents":[{"commit":"b95a89e3ff65f175b0b4b422ffa968d676adacbf","subject":"Fix get_endpoint_data for non-keystone plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/b95a89e3ff65f175b0b4b422ffa968d676adacbf"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-27 16:58:44.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-28 14:48:08.000000000","tz":120},"subject":"Correct endpoint and major version discovery for non-keystone plugins","message":"Correct endpoint and major version discovery for non-keystone plugins\n\nWhen a non-keystone plugin is used together with an unversioned endpoint,\nwe give up on discovery before figuring out both major version and\nthe correct endpoint.\n\nChange-Id: Ie623931b150748d7759cf276e0023a2f06a8d4db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/5df761193e8333f0029627d23ac88904c5c28b08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/5df761193e8333f0029627d23ac88904c5c28b08"}]},"branch":"refs/heads/master"},"981a19bba1d5a8477d51ed4751c87a3a4d130aee":{"kind":"REWORK","_number":4,"created":"2020-07-31 09:32:41.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/82/743282/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/82/743282/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/82/743282/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/82/743282/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/82/743282/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/82/743282/4"}}},"commit":{"parents":[{"commit":"b95a89e3ff65f175b0b4b422ffa968d676adacbf","subject":"Fix get_endpoint_data for non-keystone plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/b95a89e3ff65f175b0b4b422ffa968d676adacbf"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-27 16:58:44.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-31 09:32:26.000000000","tz":120},"subject":"Correct major version discovery for non-keystone plugins","message":"Correct major version discovery for non-keystone plugins\n\nWhen a non-keystone plugin is used together with an unversioned endpoint,\nwe give up on discovery before figuring out both major version and\nthe correct endpoint. This is because get_endpoint_data is called with\ndiscover_versions\u003dFalse, so discovery assumes we have all information\nalready. It may be an issue in discovery itself, but I\u0027m afraid to\ntouch that code. Instead, if get_endpoint_data returns no API version\nwith discover_versions\u003dFalse, try with discover_versions\u003dTrue, which\nmatches what the identity plugins do.\n\nAlso increase the unit test coverage.\n\nChange-Id: Ie623931b150748d7759cf276e0023a2f06a8d4db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/981a19bba1d5a8477d51ed4751c87a3a4d130aee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/981a19bba1d5a8477d51ed4751c87a3a4d130aee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
