)]}'
{"id":"openstack%2Fkeystoneauth~841169","triplet_id":"openstack%2Fkeystoneauth~master~I4969a42ee651ac2c559e378d879b673a1d788c57","project":"openstack/keystoneauth","branch":"master","attention_set":{},"removed_from_attention_set":{"16465":{"account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"last_update":"2023-10-06 17:25:18.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2023-10-06 17:25:18.000000000","reason":"Change was submitted"},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2023-10-06 17:25:18.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2023-10-06 17:25:18.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2023-10-06 17:25:18.000000000","reason":"Change was submitted"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2023-10-06 17:25:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4969a42ee651ac2c559e378d879b673a1d788c57","subject":"Drop parameters when connecting to a redirected endpoint","status":"MERGED","created":"2022-05-09 21:09:18.000000000","updated":"2023-10-06 17:26:33.000000000","submitted":"2023-10-06 17:25:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841169","meta_rev_id":"c4f04fcbcc6c134b49afb8629856731d54f41b0c","_number":841169,"virtual_id_number":841169,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-06 17:25:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2023-10-06 14:22:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2023-10-06 14:22:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2023-10-06 14:22:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":31863,"name":"Grzegorz Grasza","email":"grzegorz@grasza.com"}],"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-09 22:58:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-10 07:59:35.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2022-09-12 19:08:58.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"},{"updated":"2023-05-11 15:23:54.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"CC"},{"updated":"2023-05-11 15:23:54.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"CC"},{"updated":"2023-05-11 15:23:54.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":31863,"name":"Grzegorz Grasza","email":"grzegorz@grasza.com"},"state":"CC"},{"updated":"2023-05-11 15:23:54.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2023-09-16 16:04:05.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2023-10-06 14:22:15.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2023-10-06 14:22:29.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"d72ca49c64771decbe4ab5e2a7d6adc8719624da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-05-09 21:09:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d55bd26823e9a2f33c9a0b80baa3a393f6743610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-09 22:58:11.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/67aaa98871b14a109e2b4d989b1b3303 : SUCCESS in 1h 47m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc81b5d4d51f42b78aa5bff947c4f2b4 : FAILURE in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c342c6cf904477da5a7c01bf24933fe : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ffac9a9afdf840f9ab1724d694b72641 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8fecc0847e640fb8e4873f167f4ef0a : SUCCESS in 3m 23s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/03d93529a9d743f7ada3e7969e8e01aa : FAILURE in 1h 00m 01s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/dc0087d4868844e583686d39b1419fa8 : SUCCESS in 4m 00s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/48d4549b19b042e2883601f5e36fb4c6 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9de29255c9e4678b6e1649d874f1e19 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"31a07293a270e6487332b487012972b6f6b31da1","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-05-10 07:59:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e7f4d04aad3ba0e4ca0d7e79fe5214f57ec074b","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-05-10 13:48:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b19fcbc8e66bb29140874397db0c8515bcab21ad","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-05-10 13:50:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f939bef7cb786c0335403cf27c306fef31a2d1a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-05-10 14:13:37.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"09a72f3ba5f11c68d21e78fa1257d9d675e93993","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 15:28:17.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/017278d6146f4441bc1c1d4068b1a715 : SUCCESS in 1h 12m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5410842eb6647ffbadd07c5ebd56209 : FAILURE in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e50279e33edc439a84cb8a688b0f93dd : SUCCESS in 4m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d3bd8887b9d74f3eb192b125308eefac : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5e01b368ca248fc9e3c0a57d8003102 : SUCCESS in 4m 02s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/bab9b503a3a34633b9bf689aa83e0444 : SUCCESS in 34m 01s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/db38092b30604c4c947cecd3f0d9d543 : SUCCESS in 4m 43s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/3bd974eee4ff4631ad131ec66421a643 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c6ca25c5aab407280745e21903101fb : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":2},{"id":"1e67232869655d1b42a0c3d70dc2266a1e31994a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-05-11 21:17:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"439fbdc0bb79c4cbe9279fe074524847e73c5671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 23:04:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2cdc6949cdf64bdb8d9878e1324d1fc8 : SUCCESS in 1h 40m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5b9bc4d496a4de8bbb8fe4f784dca87 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d9f4d6c25d84e8488d3625928a94832 : FAILURE in 3m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8937ede89af64b2ca6eef40152868f62 : FAILURE in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ddc6399d201453f9f2c286a52b5953e : FAILURE in 4m 22s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a60dd5881d00481ba42e3a28985dcc9a : SUCCESS in 46m 12s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/5df95b3b89a7404989bbfe8ff30bb3da : SUCCESS in 4m 42s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/e8a3b40ccc9841e4a741f2ac22e783ae : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef96974cfc034160a60486a3b69d34b7 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b230bc6f954443091720ce999c5a64f : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":3},{"id":"be1cbe231a762dcb3528074f23236f066cbb20e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-05-12 01:31:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4a5ff0ca0719ff68171ea35037306848409da1fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 02:57:42.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8578954b331e4525842ebd7a6379c1b1 : SUCCESS in 1h 25m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/404dea7d911944b78f8eaed8ca372d38 : FAILURE in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3748c5dedd3546ccae8ccf3e35324f8a : FAILURE in 3m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/26686b8db0f84c91a2b276a2806582c6 : FAILURE in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd3906c23a6d45e4b7ddd5ef33815a66 : FAILURE in 3m 17s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9dbcb90a8c134689b69da7ab880976ff : SUCCESS in 48m 11s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/7226b0058785447693b171fd641a0e98 : SUCCESS in 4m 19s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/4be677c8e3ce4ef29052f922da2ae8d0 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9327f66670574f19997f9e7a5e043e54 : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/052936651ec8433c81f4f3ff7d9f4544 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":4},{"id":"a12a0e26d05c2a1a5d608923c9c7ae9139a297d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-05-13 12:50:31.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"12a05b49d0434b23f2ab7dcfa2808f323ce0896e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 14:15:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c5038386dbb469d9a85c6664bd1f567 : SUCCESS in 1h 22m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c1b48d06c784074a5d37f1525ee8a56 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90943d6c90784c2e9288bb93afb66b82 : FAILURE in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29422f707e584dac9af71e3285984603 : FAILURE in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/212b7d0406eb473d930ad0d223dd0f29 : FAILURE in 4m 33s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/2079d108563a4d039dd7d59e326a5838 : SUCCESS in 47m 11s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/0f34351ca1544bd1a53c4ca98e5150d6 : SUCCESS in 4m 57s\n- osc-tox-py36-tips https://zuul.opendev.org/t/openstack/build/e5011940d6004d4383eddd45bec47f7e : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/314f6595b79d499c9e199e54d08c90e8 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8377b2e91c254ce5900da86aafeb472a : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":5},{"id":"5d1a4f724206b66b80ef18e41834373e48e027c6","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-07-20 20:34:27.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"2ab22ab598f237cc1f75664989a0932eecb007d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 22:20:53.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bdb57b1e478f41dcb9abf16b929f828f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5dad6953ec9d4da39a14837bf447d133 : SUCCESS in 1h 44m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8dc9c40cd6543da829e95c36e051a10 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a6254a2adcb41b9842b81de667d19ce : FAILURE in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da17563c2c414ec19c040f3cd45960ca : FAILURE in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc5b4b927b9f4ad6b23005367a3ebc30 : RETRY_LIMIT in 3m 01s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e17c795ca484481eac9fd6eeeb148080 : SUCCESS in 1h 07m 02s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/70ad7fbeaa9d4a80bb829f65d7f7f938 : SUCCESS in 8m 09s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/7f8eac9519884910b8faaa577672622a : SUCCESS in 9m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f354e6bd00647a6b776aa6a2723620e : SUCCESS in 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c20114ad72d48228068c64c3575a73d : SUCCESS in 7m 23s","accounts_in_message":[],"_revision_number":5},{"id":"7383544be204dcdaefc050cb4cf52a199147ea6d","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-09-12 19:08:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8e606f129eb67fc90a857565ba2d77742d7c3a8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 20:36:47.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/681ee2bdb985454ab0df80f0b30b4d01\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5050fc27c7c04448b34038cf09928098 : SUCCESS in 1h 26m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f64983f6363040c3a022aeb474d29395 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c75156d872a463d8a9fe95cda4927fa : FAILURE in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5ec6f9328b94fe9ba8eccfeba95bff9 : FAILURE in 6m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27db69b093cd4bcab985625c9caab20f : RETRY_LIMIT in 2m 34s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/95793e494ebc4816bb67fd3c50548f9e : SUCCESS in 1h 19m 00s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/75fd8ce756f24da6a6796b469b4e9145 : SUCCESS in 4m 43s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d07c3ed2337a4f3ba8cb45c5e73fffc0 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/789692f636a941ef9d6b0ee7e64f8de6 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a41075b69f14db69a8cfa63ef5db0b0 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":5},{"id":"89055777a224f5bdc8cb2e7329526b3c480cf425","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-09-16 21:19:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5dc8ffa7e38ade0da16ba7f32ffdf81de60fab62","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-09-16 21:20:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ae7baf880cf76efc410dd7d681bbdcc491f00a2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 23:00:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0aced07880b4cf39f05ee7021cc982f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/742e5585d18a4c60b189c5d5c08d9ea8 : SUCCESS in 1h 39m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf946b1edd743938358849dcd0c8d61 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ec7bd51cac347d293dabcffbb5802dc : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edd50d142d0a4e85bb9c794baefc4622 : SUCCESS in 6m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eac58999939a42f88cb169d3ddc421cf : RETRY_LIMIT in 2m 39s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6c9c3377b2f34f0ea567cb89bbe3188a : SUCCESS in 1h 05m 11s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d457ebdabb6146c9a97c05b6cd034f07 : SUCCESS in 4m 44s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/e7f7a15ea515462c957b9ffa373acdaf : SUCCESS in 8m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/310f96b443904fcb86dec94906e44ad5 : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d87ab0c86c314e86a32824802244accf : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":6},{"id":"00dbb0f23a1de041f4202c74dd1e5bdfbc49bd1d","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-09-20 19:00:30.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8f5a00ed7a3a33a34d7c5e51febb080a08d74ea7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-09-27 15:12:56.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"d17a7ab338bbbf0190cd300e0ffc7e6b000f9e92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 16:49:49.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17491334e76749fa9a7bb4011a75f88d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aacc47821084422dbc3c78dbe5b4aa59 : SUCCESS in 1h 34m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9d56733d17c482183c2e5f2c132fe05 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6eb774c163b04f4b8aecefcb9f9a29e6 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67949b8af8454fd29476fef0eac0334d : SUCCESS in 7m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1e4de8307b646bd848acba36aaf3a74 : RETRY_LIMIT in 1m 49s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e6720783568c4e6a872af61946092941 : SUCCESS in 54m 32s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/2b7f6b2571f9447f91d3131fadbc7610 : SUCCESS in 5m 29s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/58a4ac291f5c4cd8b3ebfadb6b545c5b : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff759c8593164319ae01be920e34ba17 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/111a494ca468431688f49e2e8cbf5f7b : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":7},{"id":"f25fefdbbea50c0f3aa7b16f58c6249938f5bb1e","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2023-05-11 15:23:54.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9e4ec4b5618bbd4e49b936bff3de0b04cf670f78","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-09-16 16:04:05.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"fee03da7582c536b55425d73004a3097ebbc4f67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-10-06 14:21:13.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"dca26fe613bd3727a6a2ca127a55b2fb7fcbb1b8","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-10-06 14:22:15.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ea27b7a1ad2fa160eb9522aa3d01c02cd586e9de","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-10-06 14:22:29.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"6e2a9100f46b99074a24134b55088151539a53cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 15:51:54.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be50a56c6e21493ea480f3d6b98d14c0\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0bea4d76db8424b9aa1ba900982d6ab : SUCCESS in 1h 27m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eb1726a022748d8a34e7f50d3363142 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f481a5974fc4ff9a6f58a7bfd0601e8 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/057f35d44c3c418fbe4582a07491ee9c : SUCCESS in 6m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f27642b97fc342d0b682924f419dc83e : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/130bc3a5c6d64a91b73640a24c8d6a99 : SUCCESS in 6m 41s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/babaacb4ae0f459d9c17a2ca16c299a7 : SUCCESS in 1h 03m 16s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d60b026ca0cb4050b26d5a421c434d3b : SUCCESS in 5m 02s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/393b45eb9dea42e7b7a4ed0fba1e27f8 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0af39ac5256a4a26a217633b3151c41f : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2785e4b8b85742009bc5d3d1f58f2505 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":8},{"id":"653aa61e6e177397ceb7583cf1db800b81ccb346","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 15:52:27.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"7348a7d445be0e6320eb438239e2f4f37ab94c88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 17:25:18.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d1f925001214120b957084726076081\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4da349c155444a81bc5aa1e6c358d3b7 : SUCCESS in 1h 32m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cc3d75fed44425aa48b9f23cd657f0b : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d0cefa554274f22bfb4947a88572e1f : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78050800d3f44e9a8979bac90fd7ad86 : SUCCESS in 8m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53985406d3584f6fb61e8205e69109b7 : SUCCESS in 3m 07s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/663308e80b544143a65674ef7b79c734 : SUCCESS in 1h 05m 41s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/df3a2003d78e4c2eb13967bcd011e5ae : SUCCESS in 5m 09s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/f5b006e595914369936f1352b68b3df3 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8545cae5cad42c19cb5f742e1522841 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bc649edbd754edeb8170ac87a49522c : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":8},{"id":"206bbcc320ccf0792a230413a1ba9d2006f1ed8d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 17:25:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"c4f04fcbcc6c134b49afb8629856731d54f41b0c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 17:26:33.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68601ff1292b4535960d17965dcd26ac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d5458f0c09e4ae8bde1c4e1187c00bb : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28c589fdb3bb49a0b4059385313ed515 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"74f05249d1df2cdfe7836c7ef096f864d18d7201","revisions":{"736935582b349dde0c72b43da7e6e8b66b600aa6":{"kind":"REWORK","_number":1,"created":"2022-05-09 21:09:18.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/69/841169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/69/841169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/1"}}},"commit":{"parents":[{"commit":"cc58eca84fb9e3ba975adb3bb4529c10dcbddbde","subject":"Merge \"Improve help messages for a few options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/cc58eca84fb9e3ba975adb3bb4529c10dcbddbde"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-05-09 21:05:38.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-05-09 21:06:01.000000000","tz":-420},"subject":"WIP: Only include parameters not present upon redirect","message":"WIP: Only include parameters not present upon redirect\n\nWhen redirected, the server *generally* returns a fully\nformed URI, but does not really have to, so we may end up\nin a \"depending on how the redirect was triggered\" would\nresult in the response handling.\n\nWhen a fully known endpoint is redirected, before this fix\nthe parameters would be appended to the command line twice\nas it would become a merge of outstanding parameters and the\nlocation uri provided by the remote server.\n\nExample of what was occuring, when only provision_state\u003dactive\nwas a parameter before the redirect:\n\n/v1/nodes?provision_state\u003dactive\u0026provision_state\u003dactive\n\nStory: 2010029\nTask: 45316\nChange-Id: I4969a42ee651ac2c559e378d879b673a1d788c57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/736935582b349dde0c72b43da7e6e8b66b600aa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/736935582b349dde0c72b43da7e6e8b66b600aa6"}]},"branch":"refs/heads/master"},"717bd78ae746b591d429d6e634fdc83633587baf":{"kind":"REWORK","_number":2,"created":"2022-05-10 14:13:37.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/69/841169/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/69/841169/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/2"}}},"commit":{"parents":[{"commit":"cc58eca84fb9e3ba975adb3bb4529c10dcbddbde","subject":"Merge \"Improve help messages for a few options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/cc58eca84fb9e3ba975adb3bb4529c10dcbddbde"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-05-09 21:05:38.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-05-10 14:13:37.000000000","tz":0},"subject":"WIP: Only include parameters not present upon redirect","message":"WIP: Only include parameters not present upon redirect\n\nWhen redirected, the server *generally* returns a fully\nformed URI, but does not really have to, so we may end up\nin a \"depending on how the redirect was triggered\" would\nresult in the response handling.\n\nWhen a fully known endpoint is redirected, before this fix\nthe parameters would be appended to the command line twice\nas it would become a merge of outstanding parameters and the\nlocation uri provided by the remote server.\n\nExample of what was occuring, when only provision_state\u003dactive\nwas a parameter before the redirect:\n\n/v1/nodes?provision_state\u003dactive\u0026provision_state\u003dactive\n\nStory: 2010029\nTask: 45316\nChange-Id: I4969a42ee651ac2c559e378d879b673a1d788c57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/717bd78ae746b591d429d6e634fdc83633587baf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/717bd78ae746b591d429d6e634fdc83633587baf"}]},"branch":"refs/heads/master"},"fda4aaae4218385a1ccd30847f7997664f209287":{"kind":"REWORK","_number":3,"created":"2022-05-11 21:17:47.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/69/841169/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/69/841169/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/3"}}},"commit":{"parents":[{"commit":"cc58eca84fb9e3ba975adb3bb4529c10dcbddbde","subject":"Merge \"Improve help messages for a few options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/cc58eca84fb9e3ba975adb3bb4529c10dcbddbde"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-05-09 21:05:38.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-05-11 21:15:44.000000000","tz":-420},"subject":"Drop parameters when connecting to a redirected endpoint","message":"Drop parameters when connecting to a redirected endpoint\n\nWhen redirected, the server *generally* returns a fully\nformed URI, but does not really have to, so we may end up\nin a \"depending on how the redirect was triggered\" would\nresult in the response handling.\n\nUltimately, any behavior which is not an fully formed URI\nwould be invalid.\n\nBut our code was taking the URI we got back, and would then\nre-issue the request with a list of parameters with the new\nURL. Duplicating the parameters on the URI.\n\nExample of what was occuring, when only provision_state\u003dactive\nwas a parameter before the redirect:\n\n/v1/nodes?provision_state\u003dactive\u0026provision_state\u003dactive\n\nStory: 2010029\nTask: 45316\nChange-Id: I4969a42ee651ac2c559e378d879b673a1d788c57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/fda4aaae4218385a1ccd30847f7997664f209287"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/fda4aaae4218385a1ccd30847f7997664f209287"}]},"branch":"refs/heads/master"},"52dbb91c10303416c636e7d06e8c875948c052ef":{"kind":"REWORK","_number":4,"created":"2022-05-12 01:31:15.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/69/841169/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/69/841169/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/4"}}},"commit":{"parents":[{"commit":"cc58eca84fb9e3ba975adb3bb4529c10dcbddbde","subject":"Merge \"Improve help messages for a few options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/cc58eca84fb9e3ba975adb3bb4529c10dcbddbde"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-05-09 21:05:38.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-05-12 01:31:09.000000000","tz":-420},"subject":"Drop parameters when connecting to a redirected endpoint","message":"Drop parameters when connecting to a redirected endpoint\n\nWhen redirected, the server *generally* returns a fully\nformed URI, but does not really have to, so we may end up\nin a \"depending on how the redirect was triggered\" would\nresult in the response handling.\n\nUltimately, any behavior which is not an fully formed URI\nwould be invalid.\n\nBut our code was taking the URI we got back, and would then\nre-issue the request with a list of parameters with the new\nURL. Duplicating the parameters on the URI.\n\nExample of what was occuring, when only provision_state\u003dactive\nwas a parameter before the redirect:\n\n/v1/nodes?provision_state\u003dactive\u0026provision_state\u003dactive\n\nStory: 2010029\nTask: 45316\nChange-Id: I4969a42ee651ac2c559e378d879b673a1d788c57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/52dbb91c10303416c636e7d06e8c875948c052ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/52dbb91c10303416c636e7d06e8c875948c052ef"}]},"branch":"refs/heads/master"},"3292513f90d18fbb6bfd361273eb1bfc45c7b7a5":{"kind":"REWORK","_number":5,"created":"2022-05-13 12:50:31.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/69/841169/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/69/841169/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/5"}}},"commit":{"parents":[{"commit":"cc58eca84fb9e3ba975adb3bb4529c10dcbddbde","subject":"Merge \"Improve help messages for a few options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/cc58eca84fb9e3ba975adb3bb4529c10dcbddbde"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-05-09 21:05:38.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-05-13 12:50:31.000000000","tz":0},"subject":"Drop parameters when connecting to a redirected endpoint","message":"Drop parameters when connecting to a redirected endpoint\n\nWhen redirected, the server *generally* returns a fully\nformed URI, but does not really have to, so we may end up\nin a \"depending on how the redirect was triggered\" would\nresult in the response handling.\n\nUltimately, any behavior which is not an fully formed URI\nwould be invalid.\n\nBut our code was taking the URI we got back, and would then\nre-issue the request with a list of parameters with the new\nURL. Duplicating the parameters on the URI.\n\nExample of what was occuring, when only provision_state\u003dactive\nwas a parameter before the redirect:\n\n/v1/nodes?provision_state\u003dactive\u0026provision_state\u003dactive\n\nStory: 2010029\nTask: 45316\nChange-Id: I4969a42ee651ac2c559e378d879b673a1d788c57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/3292513f90d18fbb6bfd361273eb1bfc45c7b7a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/3292513f90d18fbb6bfd361273eb1bfc45c7b7a5"}]},"branch":"refs/heads/master"},"689dd7a8e2d30daeba6dd94b18e15d61eaf9bdea":{"kind":"REWORK","_number":6,"created":"2022-09-16 21:19:28.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/69/841169/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/69/841169/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/6"}}},"commit":{"parents":[{"commit":"cc58eca84fb9e3ba975adb3bb4529c10dcbddbde","subject":"Merge \"Improve help messages for a few options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/cc58eca84fb9e3ba975adb3bb4529c10dcbddbde"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-05-09 21:05:38.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2022-09-16 21:19:26.000000000","tz":-420},"subject":"Drop parameters when connecting to a redirected endpoint","message":"Drop parameters when connecting to a redirected endpoint\n\nWhen redirected, the server *generally* returns a fully\nformed URI, but does not really have to, so we may end up\nin a \"depending on how the redirect was triggered\" would\nresult in the response handling.\n\nUltimately, any behavior which is not an fully formed URI\nwould be invalid.\n\nBut our code was taking the URI we got back, and would then\nre-issue the request with a list of parameters with the new\nURL. Duplicating the parameters on the URI.\n\nExample of what was occuring, when only provision_state\u003dactive\nwas a parameter before the redirect:\n\n/v1/nodes?provision_state\u003dactive\u0026provision_state\u003dactive\n\nStory: 2010029\nTask: 45316\nChange-Id: I4969a42ee651ac2c559e378d879b673a1d788c57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/689dd7a8e2d30daeba6dd94b18e15d61eaf9bdea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/689dd7a8e2d30daeba6dd94b18e15d61eaf9bdea"}]},"branch":"refs/heads/master"},"ddcafb04c0f703001c469fa14530e23a5acb54ce":{"kind":"REWORK","_number":7,"created":"2022-09-27 15:12:56.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/69/841169/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/69/841169/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/7"}}},"commit":{"parents":[{"commit":"cc58eca84fb9e3ba975adb3bb4529c10dcbddbde","subject":"Merge \"Improve help messages for a few options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/cc58eca84fb9e3ba975adb3bb4529c10dcbddbde"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-05-09 21:05:38.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2022-09-27 15:12:21.000000000","tz":-420},"subject":"Drop parameters when connecting to a redirected endpoint","message":"Drop parameters when connecting to a redirected endpoint\n\nWhen redirected, the server *generally* returns a fully\nformed URI, but does not really have to, so we may end up\nin a \"depending on how the redirect was triggered\" would\nresult in the response handling.\n\nUltimately, any behavior which is not an fully formed URI\nwould be invalid.\n\nBut our code was taking the URI we got back, and would then\nre-issue the request with a list of parameters with the new\nURL. Duplicating the parameters on the URI.\n\nExample of what was occuring, when only provision_state\u003dactive\nwas a parameter before the redirect:\n\n/v1/nodes?provision_state\u003dactive\u0026provision_state\u003dactive\n\nCo-Authored-By: Kristi Nikolla \u003cknikolla@bu.edu\u003e\nCo-Authored-By: Jay Faulkner \u003cjay@jvf.cc\u003e\n\nStory: 2010029\nTask: 45316\nChange-Id: I4969a42ee651ac2c559e378d879b673a1d788c57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/ddcafb04c0f703001c469fa14530e23a5acb54ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/ddcafb04c0f703001c469fa14530e23a5acb54ce"}]},"branch":"refs/heads/master"},"74f05249d1df2cdfe7836c7ef096f864d18d7201":{"kind":"REWORK","_number":8,"created":"2023-10-06 14:21:13.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/69/841169/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/69/841169/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/69/841169/8"}}},"commit":{"parents":[{"commit":"659a288cc145db280e32d9b5c968861e6df8c51b","subject":"Merge \"Keep request-id at redirect\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/659a288cc145db280e32d9b5c968861e6df8c51b"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-05-09 21:05:38.000000000","tz":-420},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2023-10-06 14:21:02.000000000","tz":120},"subject":"Drop parameters when connecting to a redirected endpoint","message":"Drop parameters when connecting to a redirected endpoint\n\nWhen redirected, the server *generally* returns a fully\nformed URI, but does not really have to, so we may end up\nin a \"depending on how the redirect was triggered\" would\nresult in the response handling.\n\nUltimately, any behavior which is not an fully formed URI\nwould be invalid.\n\nBut our code was taking the URI we got back, and would then\nre-issue the request with a list of parameters with the new\nURL. Duplicating the parameters on the URI.\n\nExample of what was occuring, when only provision_state\u003dactive\nwas a parameter before the redirect:\n\n/v1/nodes?provision_state\u003dactive\u0026provision_state\u003dactive\n\nCo-Authored-By: Kristi Nikolla \u003cknikolla@bu.edu\u003e\nCo-Authored-By: Jay Faulkner \u003cjay@jvf.cc\u003e\n\nStory: 2010029\nTask: 45316\nChange-Id: I4969a42ee651ac2c559e378d879b673a1d788c57\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/74f05249d1df2cdfe7836c7ef096f864d18d7201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/74f05249d1df2cdfe7836c7ef096f864d18d7201"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
