)]}'
{"id":"openstack%2Fkeystoneauth~877723","triplet_id":"openstack%2Fkeystoneauth~master~Ibdbc0b9228c6b24a521b9117ee0298a4a302bb98","project":"openstack/keystoneauth","branch":"master","topic":"whl-metadata","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2023-10-06 15:28:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibdbc0b9228c6b24a521b9117ee0298a4a302bb98","subject":"Add project URLs from README as packaging metadata","status":"MERGED","created":"2023-03-16 23:04:23.000000000","updated":"2023-10-06 15:30:19.000000000","submitted":"2023-10-06 15:28:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877723-whl-metadata","meta_rev_id":"fc0d2dcb700fe0f3675c9d4cf315dce39f10e47c","_number":877723,"virtual_id_number":877723,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-06 15:28:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":2,"date":"2023-10-06 14:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":2,"date":"2023-09-16 16:12:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":1,"date":"2023-10-06 14:09:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-17 00:18:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-16 16:12:03.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2023-10-06 14:09:51.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"dd7f281a7522bbd27c82dcb8f3d7cb65ea896e48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-16 23:04:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"858c1d218a7ff593ae65bc0dbf89d5bde87eed95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 00:18:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bd4ee68792b46af9ecacb97502ec3b5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e1ab03bd3eee4f6c964abb088322279c : SUCCESS in 2m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8f7128adb9448d390ed9054830c6e75 : SUCCESS in 1h 07m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd9ed14d688944509cfa48566a15b0b0 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9fcd6db85a841738682c3f5788dab9d : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b4213bc9308483fa4055b5f85f53827 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8a64b4fc1284e50b712d68859cba53b : SUCCESS in 3m 04s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/33178ee9702e4cb2a666c978464ea600 : SUCCESS in 44m 19s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/04619d7cd493427b8547ad16e75597b1 : FAILURE in 4m 03s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/ab3b231cc19942fabec329e3e1c71ca7 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10aae7c2809243e29740856248b53715 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f47b45f6a6dd7da2a606966ce7fe25e1f42fb3d4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-24 16:00:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af3a8cb83d94d8e0f45c0153a8dd9fe7b0f7b4c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-24 17:54:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed2d4f839de348ecb4767b75bb4493bd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a984250200a143f8b2286217d3395723 : SUCCESS in 1m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50013a3fdc2444d4bce951f542a4bd91 : SUCCESS in 1h 45m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57b58eb1e6334179ab4b1684c987387d : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d619912721414c99a45fafb0bc2f575b : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7eb1fa73ba76467c9cc566e0e72d16ef : SUCCESS in 7m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22cd4b9be7524e5bbc6545149adaa150 : SUCCESS in 3m 22s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/e3e859cf9a274f85af5913eeb472e7b3 : SUCCESS in 1h 08m 20s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/5b19a8507d0c400bae5c6eeabc9a7d00 : SUCCESS in 4m 36s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/1ddeeb5febfc40a2870eeb38b3b18052 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8768e6b22d1c4a6fbe0b2d08275ba4b0 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"73334677c6b7286cf42bfa0556f7825fe0cdbd7d","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-09-16 16:12:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9f4a48c771d90389c62231d088e63310f7971c4c","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-10-06 14:09:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f9ce6fc2464993f750d9d7064ccc290b8014826b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 14:10:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6cb16ac0fb9f621c7c8e45da2490b5cb3fda5432","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 15:28:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33a5975e0d9d433a944ae37681cb9376\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/48276cf0c1284caa90b2c249fd497ba1 : SUCCESS in 2m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ca4006531e3438e88d1a3b0a94f57e3 : SUCCESS in 1h 16m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f309ad22a012497caacc95729ad28295 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26d29af24a6445cd99e665d6e6c9c771 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c77c342abef44cc3b96cf7132f6d7a39 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc9e6bfc09db4a1380957fd2a88b7b68 : SUCCESS in 3m 49s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/09e96ea4c4ca440992b7bf8c51ef1ab8 : SUCCESS in 1h 12m 06s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/f03d6a0815a6483daa1ecc9df2a7f884 : SUCCESS in 4m 26s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/de69e4a479c4440aac379b0d6b922bad : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22b80b9313214c1889be3a227a0f0a5a : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"95b1ab65981441cbbcb203a6c6c948a5de1c1628","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 15:28:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fc0d2dcb700fe0f3675c9d4cf315dce39f10e47c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 15:30:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/676fc74a47464850a080edd9fce7052e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96abf75585b14090950f7d4891dce5d4 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66cd04e2ab5ee3f1a87b4c806337fcf2fad1287d","revisions":{"66cd04e2ab5ee3f1a87b4c806337fcf2fad1287d":{"kind":"REWORK","_number":1,"created":"2023-03-16 23:04:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/23/877723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/23/877723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/23/877723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/23/877723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/23/877723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/23/877723/1"}}},"commit":{"parents":[{"commit":"4e518145c332299af3fe921255dfdee694f96945","subject":"Fix up some packaging metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/4e518145c332299af3fe921255dfdee694f96945"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-03-16 23:03:26.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-03-16 23:03:30.000000000","tz":-420},"subject":"Add project URLs from README as packaging metadata","message":"Add project URLs from README as packaging metadata\n\nChange-Id: Ibdbc0b9228c6b24a521b9117ee0298a4a302bb98\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/66cd04e2ab5ee3f1a87b4c806337fcf2fad1287d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/66cd04e2ab5ee3f1a87b4c806337fcf2fad1287d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
