)]}'
{"id":"openstack%2Fkeystoneauth~883925","triplet_id":"openstack%2Fkeystoneauth~stable%2F2023.1~I2392ef51302804c0c66c0fb52227db5f35bca3fd","project":"openstack/keystoneauth","branch":"stable/2023.1","topic":"one_login_compat-stable/2023.1","attention_set":{},"removed_from_attention_set":{"16465":{"account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"last_update":"2023-05-25 15:24:41.000000000","reason":"\u003cGERRIT_ACCOUNT_16465\u003e replied on the change","reason_account":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2023-05-25 15:19:16.000000000","reason":"\u003cGERRIT_ACCOUNT_7973\u003e replied on the change","reason_account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2023-05-25 16:47:03.000000000","reason":"Change was submitted"},"35976":{"account":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"last_update":"2023-05-25 16:47:03.000000000","reason":"Change was submitted"},"782":{"account":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"last_update":"2023-05-25 16:47:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2392ef51302804c0c66c0fb52227db5f35bca3fd","subject":"Make v3oidcpassword send client_id","status":"MERGED","created":"2023-05-24 09:02:00.000000000","updated":"2023-05-25 16:48:03.000000000","submitted":"2023-05-25 16:47:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883925-one_login_compat-stable/2023.1","cherry_pick_of_change":881969,"cherry_pick_of_patch_set":2,"meta_rev_id":"eac8b0e023ff7616630aa1e8c6eab28b43a69bc5","_number":883925,"virtual_id_number":883925,"owner":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-25 16:47:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2023-05-25 15:24:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2023-05-25 15:19:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":1,"date":"2023-05-25 15:24:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"reviewers":{"CC":[{"_account_id":26922,"name":"Steven Relf","email":"s.relf@eschercloud.ai","username":"srelf"}],"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-24 09:02:00.000000000","updated_by":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"reviewer":{"_account_id":26922,"name":"Steven Relf","email":"s.relf@eschercloud.ai","username":"srelf"},"state":"CC"},{"updated":"2023-05-24 09:05:03.000000000","updated_by":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2023-05-24 09:05:03.000000000","updated_by":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2023-05-24 09:05:03.000000000","updated_by":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2023-05-24 09:05:03.000000000","updated_by":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2023-05-24 10:30:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d0f435799e1fa2bbd1df31ab76734ab96fa9fa66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"date":"2023-05-24 09:02:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ca026a68d4bd7e919739ff9819a7736d6ca250b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 10:30:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/414c6819792341e2a4218e01319302d1\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f624fccf7d7f46a5b3bd8bac6b302234 : SUCCESS in 1h 22m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ac485a61a3c4d80a616321323af2877 : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6395b649889d4a968a7e23c7f9791d99 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12b2b0564fb64ddb866f234c74deff27 : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e080f64c60a84f53a43db49698b37b10 : SUCCESS in 3m 22s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c2719b9816894a8994bb002d23ff4fa3 : SUCCESS in 58m 42s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/7f293245021f4624820678a5c26b63e0 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e1a315c49c489fb2ecc9ed5174bf72 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"b1149ec72b4d5213d21c530156f7fdc6d7389d97","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2023-05-25 15:19:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9d88f3c8cbae06cc28dfa388a2ec74d3538dab2d","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2023-05-25 15:24:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6da5e8988be9d13a45274a326105efbe885435f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 15:25:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c1f71703eaf90addbbc1eda11ab469577bf14a1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 16:47:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ee2717acfdc4e729dd989a319d86acb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b24c5a9b2af24d2b8d13519b685cfde5 : SUCCESS in 1h 16m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f9e8ad97eda43d8a7ce8269bfb32343 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfdd157a08dd4d2aa8af7250cc21be6d : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b384c3fbae6e4cfe973a44d8bd12d2b6 : SUCCESS in 6m 43s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/7c73b2d668204db08f58655bc4364237 : SUCCESS in 1h 03m 36s\n- openstacksdk-tox-py38-tips https://zuul.opendev.org/t/openstack/build/d0d80d336e3949deacc49213be48a8b6 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd631275d8bf4e7fa8a076273e3182d9 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c83a640e37b2938ae24d67dc375e4ada6e42d14b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 16:47:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"eac8b0e023ff7616630aa1e8c6eab28b43a69bc5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 16:48:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a214407543b44df901b152eb2c1a6db\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55103f8ad9384cfcaa2ab5353164bba9 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e96c503154ff8f3560b1601ee4616549d8c5b161","revisions":{"e96c503154ff8f3560b1601ee4616549d8c5b161":{"kind":"REWORK","_number":1,"created":"2023-05-24 09:02:00.000000000","uploader":{"_account_id":35976,"name":"Graeme Moss","display_name":"Graeme Moss","email":"graeme.moss@nscale.com","username":"gmoss"},"ref":"refs/changes/25/883925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/25/883925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/25/883925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/25/883925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/25/883925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/25/883925/1"}}},"commit":{"parents":[{"commit":"97c2d0b95fa5b512d74c23ee21d7e8c9ae11b0df","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/97c2d0b95fa5b512d74c23ee21d7e8c9ae11b0df"}]}],"author":{"name":"Steve Relf","email":"s.relf@eschercloud.ai","date":"2023-05-02 14:03:35.000000000","tz":60},"committer":{"name":"Graeme Moss","email":"g.moss@eschercloud.ai","date":"2023-05-24 09:02:00.000000000","tz":0},"subject":"Make v3oidcpassword send client_id","message":"Make v3oidcpassword send client_id\n\nUpdates the v3oidcpassword function to also send the client_id as part of the request body,\nthis seems to be a requirement for services like onelogin.\n\nChange-Id: I2392ef51302804c0c66c0fb52227db5f35bca3fd\n(cherry picked from commit 30d61ad316bc3fe873674b2825bab1e85534b80d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/e96c503154ff8f3560b1601ee4616549d8c5b161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/e96c503154ff8f3560b1601ee4616549d8c5b161"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
