)]}'
{"id":"openstack%2Fkeystoneauth~888033","triplet_id":"openstack%2Fkeystoneauth~master~I7775ec31ffa79160c696805d4ae607067b1463c7","project":"openstack/keystoneauth","branch":"master","attention_set":{},"removed_from_attention_set":{"31068":{"account":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"last_update":"2023-07-10 12:46:53.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I7775ec31ffa79160c696805d4ae607067b1463c7","subject":"Replace deprecated datetime.utcnow method","status":"ABANDONED","created":"2023-07-10 11:58:58.000000000","updated":"2024-03-08 13:27:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"edda7f20b410507d7bb7bbec6b6c9ab722c424cd","_number":888033,"virtual_id_number":888033,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-07-10 13:04:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-07-10 12:31:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"025b8b2d25cf600aabf3d528cc3624727c584b46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-07-10 11:58:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ef791b01f82883b4a1948ea5d978a3cf3c3b141","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 12:31:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/66ea0d1de6e04127979878ebad89030d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/064d61e7c6e74331b7780897f47dbf08 : FAILURE in 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08c2b67deda84d3dbccc4dc19f85ec53 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15b615a4e7bf4a43a08cdce2cad4edeb : FAILURE in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84c48a332acb4b8ba03cc4ce6b1cda40 : FAILURE in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53e43b73532e4cd89044aeb914f0345f : FAILURE in 3m 12s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ac35d065202c41b8babb0ddaed52908c : POST_FAILURE in 22m 41s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/2dca0c0a63334fcc973655488b6972d5 : FAILURE in 5m 01s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/b6d743bf63da4df78217beb6d9452256 : FAILURE in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d7f3b1dac7847fe8a32b87a375fec2b : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":1},{"id":"2f04be9cac9aec2b966efb88b31ee079ef0d9413","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-07-10 12:46:53.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"a056547e45f9c3f6398040c988cbb5f3c968981d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2023-07-10 12:50:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c2f8068d51a7d7c845e887fa8376ed8b211f1e79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 13:04:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4c4669d05c644a8966facc92dc04894\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/39a58608fcc54deeaeacce285fed6435 : FAILURE in 11m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0686cbb455d641639243ebf99fc7cdc6 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33c8603d9f6d43589fa3a7529dd7e8a8 : FAILURE in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0947fb3bc90401cbaac971f85c47b8f : FAILURE in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3107a464f15d4798b226ea2214f61bee : FAILURE in 4m 33s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/1c60e6218f144ad1944d7a054d828e36 : POST_FAILURE in 11m 07s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/b4e8699dbef743fe84afde2c0db54c84 : SUCCESS in 6m 11s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/ef3f406d83114a8391204017b4cf2d97 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/051fbf5d5d9948fdbc7b17694daf97e1 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":2},{"id":"edda7f20b410507d7bb7bbec6b6c9ab722c424cd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2024-03-08 13:27:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"23b5967f1641d06e3dbb46b95215bd5817dd7c11","revisions":{"42b94097440c11d98597b1a05fdc3af6f072e594":{"kind":"REWORK","_number":1,"created":"2023-07-10 11:58:58.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/33/888033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/33/888033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/33/888033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/33/888033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/33/888033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/33/888033/1"}}},"commit":{"parents":[{"commit":"28048af9593740315df8d9027c3bd6cae5e0a715","subject":"Merge \"Add doc of OAuth 2.0 Mutual-TLS Authenticate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/28048af9593740315df8d9027c3bd6cae5e0a715"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-07-10 11:52:04.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-07-10 11:58:56.000000000","tz":120},"subject":"Replace deprecated datetime.utcnow method","message":"Replace deprecated datetime.utcnow method\n\nWhile rebuilding openstacksdk on Fedora/Python3.12 [1], we hit\nthe deprecation warning on datetime.utcnow method [2].\nThis patch updates it with the recommended way.\n\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d2220377\n[2] https://docs.python.org/3.12/library/datetime.html#datetime.datetime.utcnow\n\nChange-Id: I7775ec31ffa79160c696805d4ae607067b1463c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/42b94097440c11d98597b1a05fdc3af6f072e594"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/42b94097440c11d98597b1a05fdc3af6f072e594"}]},"branch":"refs/heads/master"},"23b5967f1641d06e3dbb46b95215bd5817dd7c11":{"kind":"REWORK","_number":2,"created":"2023-07-10 12:50:01.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/33/888033/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/33/888033/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/33/888033/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/33/888033/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/33/888033/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/33/888033/2"}}},"commit":{"parents":[{"commit":"28048af9593740315df8d9027c3bd6cae5e0a715","subject":"Merge \"Add doc of OAuth 2.0 Mutual-TLS Authenticate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/28048af9593740315df8d9027c3bd6cae5e0a715"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-07-10 11:52:04.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2023-07-10 12:49:51.000000000","tz":120},"subject":"Replace deprecated datetime.utcnow method","message":"Replace deprecated datetime.utcnow method\n\nWhile rebuilding openstacksdk on Fedora/Python3.12 [1], we hit\nthe deprecation warning on datetime.utcnow method [2].\nThis patch updates it with the recommended way.\n\n[1] https://bugzilla.redhat.com/show_bug.cgi?id\u003d2220377\n[2] https://docs.python.org/3.12/library/datetime.html#datetime.datetime.utcnow\n\nChange-Id: I7775ec31ffa79160c696805d4ae607067b1463c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/23b5967f1641d06e3dbb46b95215bd5817dd7c11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/23b5967f1641d06e3dbb46b95215bd5817dd7c11"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
