)]}'
{"id":"openstack%2Fkeystoneauth~900051","triplet_id":"openstack%2Fkeystoneauth~master~Id771077b1070a80b9acab5ffda66155f0dea22be","project":"openstack/keystoneauth","branch":"master","topic":"change_retries_log_level","attention_set":{},"removed_from_attention_set":{"29037":{"account":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"last_update":"2024-01-05 17:09:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id771077b1070a80b9acab5ffda66155f0dea22be","subject":"Change retries log level to warning instead of info","status":"MERGED","created":"2023-11-03 15:15:32.000000000","updated":"2024-01-05 18:02:26.000000000","submitted":"2024-01-05 17:09:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900051-change_retries_log_level","meta_rev_id":"16a5cc88fd820631a9ba1267a0670ed30ca29b12","_number":900051,"virtual_id_number":900051,"owner":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-05 17:09:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2023-11-08 16:01:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2024-01-05 18:02:26.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2024-01-05 15:38:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2024-01-05 15:38:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-03 16:29:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-08 16:01:07.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-01-05 15:38:56.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-01-05 18:02:26.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"654fd9992d36d3aa275f40e9be7fc7ef2a396906","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"date":"2023-11-03 15:15:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d722d5c18714e0b30158be888ee49d8a400ac68e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 16:29:07.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2bb3998d5ed45e8850716d9e26b7cf9\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/982c4f62836447afad8bf32f8989faac : FAILURE in 1h 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83e1a77958874d6e90150e36b44f6db9 : FAILURE in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cacd08b857642b9b124d7bb183cc070 : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/043d6c54f24f4d18b8e38719c3ebd5d0 : SUCCESS in 7m 38s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0346cccf985d47489cd86e984a876735 : SUCCESS in 1h 07m 09s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/1807b5141520428483f3fa0e71c9d59d : SUCCESS in 4m 37s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/ed4f226dbfbe4b0a927e0fa5a0009f32 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d615d5675340558ea8c6bffb885ed1 : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2e0d7348050eccfb7bfdde366d64a2f86fce5b13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"date":"2023-11-06 08:57:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b90e47329238028331e2513c1eced8641e442ef7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 10:39:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8507814fa2754746ad4c528127fd5c80\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/99d36f0395384c938e0248b78f3bcc94 : SUCCESS in 1h 35m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/045a1d03ad2841e896254a932c5ab782 : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c58f764bfe5f41379e8725bce017ddd1 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c048e8f9e5b64a0f8d40d6cd2e29a913 : SUCCESS in 7m 49s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/69a8ba8eff874ab9a0da3a06fccf3741 : SUCCESS in 1h 04m 03s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/2684992d55e647b3a8d96d8dee7a5847 : SUCCESS in 4m 18s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/14303a68412c4d2e92d97f366aef244c : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b92db5046fd47b8910db9453a930dc5 : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":2},{"id":"4890d71088a9c65080254694c8caa62f362d1545","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-11-08 16:01:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a93658d86bae5e9cb50dcbbcd0b441eb1590283a","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-01-05 15:38:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"108b7812ae9b2223dd5b281cf67fa1ed6aea7ee0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 15:39:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8b9554c477dc2b268183c50353fdd7e6507ddc85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 17:09:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29cbfb7aef5e4028965a88f8f167b220\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8323dda9fd1b41bfa534105fcf31c98a : SUCCESS in 1h 28m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01aa613fe77441dfbc838e9005fce1f4 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cb448a815bc4f6b8e7682161b012bda : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af8699e4d801407a869ad946080fa2cb : SUCCESS in 6m 54s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/bcc685ad02014affadc116b19b6978ff : SUCCESS in 57m 29s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/f3f61874c5e44ed9a8015e292c57868a : SUCCESS in 5m 28s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/4cdf950959b442cc87de6e1595194745 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ebca5e5bce94fd0b84f8e02ed4fdd2d : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":2},{"id":"8446136f90d88c7faa1b10db40a8501f1843a981","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 17:09:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"874c04a1e33c37d2b6f8e06f03b65e0f85dfab69","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 17:10:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12f1b51b9e974e49ba7cedb820d18cc8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30c76a362fef4d98ae43a98ebbcd71ed : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2},{"id":"16a5cc88fd820631a9ba1267a0670ed30ca29b12","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-01-05 18:02:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b43cad93539ba00c2584030273f29ae97dfd4590","revisions":{"236666598e8e2a3e99577794ab0ee8435f4bc6af":{"kind":"REWORK","_number":1,"created":"2023-11-03 15:15:32.000000000","uploader":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"ref":"refs/changes/51/900051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/51/900051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/51/900051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/51/900051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/51/900051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/51/900051/1"}}},"commit":{"parents":[{"commit":"44e477dfb5821cb535d880738a965901ecd6f734","subject":"Merge \"Drop parameters when connecting to a redirected endpoint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/44e477dfb5821cb535d880738a965901ecd6f734"}]}],"author":{"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@corp.ovh.com","date":"2023-11-03 15:06:16.000000000","tz":60},"committer":{"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@corp.ovh.com","date":"2023-11-03 15:06:16.000000000","tz":60},"subject":"Change retries log level to warning instead of info","message":"Change retries log level to warning instead of info\n\nCloses-bug: 2042670\nChange-Id: Id771077b1070a80b9acab5ffda66155f0dea22be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/236666598e8e2a3e99577794ab0ee8435f4bc6af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/236666598e8e2a3e99577794ab0ee8435f4bc6af"}]},"branch":"refs/heads/master"},"b43cad93539ba00c2584030273f29ae97dfd4590":{"kind":"REWORK","_number":2,"created":"2023-11-06 08:57:29.000000000","uploader":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"ref":"refs/changes/51/900051/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/51/900051/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/51/900051/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/51/900051/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/51/900051/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/51/900051/2"}}},"commit":{"parents":[{"commit":"44e477dfb5821cb535d880738a965901ecd6f734","subject":"Merge \"Drop parameters when connecting to a redirected endpoint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/44e477dfb5821cb535d880738a965901ecd6f734"}]}],"author":{"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@corp.ovh.com","date":"2023-11-03 15:06:16.000000000","tz":60},"committer":{"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@corp.ovh.com","date":"2023-11-06 08:25:59.000000000","tz":60},"subject":"Change retries log level to warning instead of info","message":"Change retries log level to warning instead of info\n\nCloses-bug: 2042670\nChange-Id: Id771077b1070a80b9acab5ffda66155f0dea22be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/b43cad93539ba00c2584030273f29ae97dfd4590"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/b43cad93539ba00c2584030273f29ae97dfd4590"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
