)]}'
{"id":"openstack%2Fkeystoneauth~900530","triplet_id":"openstack%2Fkeystoneauth~master~I50500c6a21bbe60050303cea4628ca9b71a3e0eb","project":"openstack/keystoneauth","branch":"master","topic":"sys-modules-iter","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2023-11-28 15:27:21.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2023-11-10 14:52:39.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-11-10 09:53:40.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2023-11-28 14:13:28.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2023-11-20 08:42:26.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},"34975":{"account":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"last_update":"2023-11-28 15:27:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I50500c6a21bbe60050303cea4628ca9b71a3e0eb","subject":"Fix \"dictionary changed size during iteration\"","status":"MERGED","created":"2023-11-09 14:23:16.000000000","updated":"2023-11-28 15:28:32.000000000","submitted":"2023-11-28 15:27:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900530-sys-modules-iter","meta_rev_id":"5c9d645781845d880b6d1b9e5fe90310c856763f","_number":900530,"virtual_id_number":900530,"owner":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-28 15:27:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"recommended":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-11-10 09:53:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-11-10 14:52:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2023-11-10 08:53:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":1,"date":"2023-11-10 09:24:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2023-11-20 08:43:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-11-28 14:13:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":2,"date":"2023-11-20 08:42:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":1,"date":"2023-11-28 14:13:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-09 15:01:11.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-11-09 17:09:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-09 18:46:34.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"CC"},{"updated":"2023-11-09 18:46:34.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-11-10 08:53:07.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2023-11-10 09:24:33.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-11-10 09:54:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2023-11-10 09:54:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2023-11-20 07:02:15.000000000","updated_by":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2023-11-20 08:43:54.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-11-28 14:11:43.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"c8329b4fca660e96a07bf8da528a2d391cc80639","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-11-09 14:23:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a71479e37b2ce38dbf1f36f1e621e01084a191e1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-09 15:01:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"941e91bfdfffd3dbb16db0f27009df197f4b36c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-11-09 15:37:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"15294cb9d4859f0ee65a600cddbf4367c559fe58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-11-09 15:38:50.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"db4a0be4c727b11873bb4bdeb75b654bc6f6ce85","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-11-09 15:40:15.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"beff713c076152321ab716a8d363ce7ee8fe6a3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:09:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c537638c91247ba8d7dad4562bcfc4c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4036c89cf6544859355b9ed69c5aa16 : SUCCESS in 1h 25m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e387da144a1347c0991c992a2f3385a4 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5382d5ce26e94b1c8b1a9c40494b45bb : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c1b583e0d884afb8979d269e1f36885 : SUCCESS in 6m 26s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6c2569f7cc6747aea2465d0b06994fd8 : SUCCESS in 55m 35s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/0ad6e28fe349413691fefff5012cd400 : SUCCESS in 7m 13s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/7656afb342ce4020801295a4645f8da9 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c93417b988045f692b43d70d642605f : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":3},{"id":"d113a423d78eb7ba3d8ee9c5e62e1a963049f0f7","author":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"date":"2023-11-09 18:46:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4d63c3e264e691cedf87ed8433896be47d5acd59","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-11-10 08:53:07.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"07219bcb6a39afcf539f5f12a7ab67f5f68476b3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-10 09:24:33.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b61d9c1f0eba607e73e7bbd255400dc9dc4c0857","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-10 09:29:39.000000000","message":"Topic set to sys-modules-iter","accounts_in_message":[],"_revision_number":3},{"id":"13a01bf5372d11a558e6b3454a3fda70f1088d29","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-10 09:53:40.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0fdd9a3e61f64f5027c37900cd8a8b8ae369a9fd","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-11-10 14:52:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"681f3e334757945b39f48efff1d8d2f6bf8199fd","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2023-11-20 08:42:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"412022af7f742aa0bc3fe388f900b1812008335f","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-11-20 08:43:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"27bee3f37e4c150089707e513ca1633672944777","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2023-11-21 14:47:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"804fd8932edf5bd1f17aaf5d47488e0664db1da2","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-11-28 14:13:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cfdedd9564a39d560ccfeca9919b18af40717e70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 14:14:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2f1e80c7e3e95ac0ad4fb1e7d61204b742c5d229","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 15:27:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52b862cd1c584f30ae1f28da74d9e41a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c7a9c379be64031a95801b1ca72a1ac : SUCCESS in 1h 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7e84b85b5ac4c3582d76fd8cdcb5e37 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce1c75baf3554192a480aac4451c3e51 : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/37d7ea777a5f40dca4b0b67d7d4f0821 : SUCCESS in 6m 15s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/58d8bd3d0fe64e3983a479f9638df9e7 : SUCCESS in 56m 42s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/00f3395f9ace403e88986c9375d42d07 : SUCCESS in 4m 39s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/2e99a13661f64aea938b36502aac822e : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/957df28f71b240279afb19af9af28dac : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":3},{"id":"41632b206f4872b98c085f8651fd8423786ca4bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 15:27:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5c9d645781845d880b6d1b9e5fe90310c856763f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 15:28:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2de111fa751e4f8c859d29b53fe57770\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f148775ef0e340518f46fba584224cf9 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3b492a7aa0bcb73c7fff62c8e7b1cd5d42118fbe","revisions":{"b8b470ba7aab3a3fef05f83bde2907b69ec6981b":{"kind":"REWORK","_number":1,"created":"2023-11-09 14:23:16.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/30/900530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/30/900530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/30/900530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/30/900530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/30/900530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/30/900530/1"}}},"commit":{"parents":[{"commit":"44e477dfb5821cb535d880738a965901ecd6f734","subject":"Merge \"Drop parameters when connecting to a redirected endpoint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/44e477dfb5821cb535d880738a965901ecd6f734"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2023-11-09 14:13:49.000000000","tz":-300},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2023-11-09 14:23:08.000000000","tz":-300},"subject":"Fix \"dictionary changed size during iteration\"","message":"Fix \"dictionary changed size during iteration\"\n\nWe have hit an error with\n\"dictionary changed size during iteration\" a few times during\nthis week in telemetry integration tests. It seems like it\u0027s\nhitting only our ubuntu based jobs, I haven\u0027t seen this error\nin a centos based job yet. Example of the failed job can be\nfound in [1], I extracted a traceback, which leads to keystoneauth1\ninto [2]. According to [3] and [4] using copy() should help with\nthe issue\n\n[1] https://zuul.opendev.org/t/openstack/build/c99db592871a441e9cddad2f4e60c2fc/console\n[2] https://paste.opendev.org/show/bpzng2EUyFh1tvBHczt7/\n[3] https://github.com/python/cpython/issues/84507\n[4] https://github.com/python/cpython/issues/89516\n\nChange-Id: I50500c6a21bbe60050303cea4628ca9b71a3e0eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/b8b470ba7aab3a3fef05f83bde2907b69ec6981b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/b8b470ba7aab3a3fef05f83bde2907b69ec6981b"}]},"branch":"refs/heads/master"},"65ccdf593c887d5e0378a10514abed2945ea56a2":{"kind":"REWORK","_number":2,"created":"2023-11-09 15:37:14.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/30/900530/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/30/900530/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/30/900530/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/30/900530/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/30/900530/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/30/900530/2"}}},"commit":{"parents":[{"commit":"44e477dfb5821cb535d880738a965901ecd6f734","subject":"Merge \"Drop parameters when connecting to a redirected endpoint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/44e477dfb5821cb535d880738a965901ecd6f734"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2023-11-09 14:13:49.000000000","tz":-300},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2023-11-09 15:36:52.000000000","tz":-300},"subject":"Fix \"dictionary changed size during iteration\"","message":"Fix \"dictionary changed size during iteration\"\n\nWe have hit an error with\n\"dictionary changed size during iteration\" a few times during\nthis week in telemetry integration tests. It seems like it\u0027s\nhitting only our ubuntu based jobs, I haven\u0027t seen this error\nin a centos based job yet. Example of the failed job can be\nfound in [1], I extracted a traceback, which leads to keystoneauth1\ninto [2]. According to [3] and [4] using copy() should help with\nthe issue\n\n[1] https://zuul.opendev.org/t/openstack/build/c99db592871a441e9cddad2f4e60c2fc/console\n[2] https://paste.opendev.org/show/bpzng2EUyFh1tvBHczt7/\n[3] https://github.com/python/cpython/issues/84507\n[4] https://github.com/python/cpython/issues/89516\n\nChange-Id: I50500c6a21bbe60050303cea4628ca9b71a3e0eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/65ccdf593c887d5e0378a10514abed2945ea56a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/65ccdf593c887d5e0378a10514abed2945ea56a2"}]},"branch":"refs/heads/master"},"3b492a7aa0bcb73c7fff62c8e7b1cd5d42118fbe":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-11-09 15:38:50.000000000","uploader":{"_account_id":34975,"name":"Jaromír Wysoglad","email":"jwysogla@redhat.com","username":"jwysogla"},"ref":"refs/changes/30/900530/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/30/900530/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/30/900530/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/30/900530/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/30/900530/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/30/900530/3"}}},"commit":{"parents":[{"commit":"44e477dfb5821cb535d880738a965901ecd6f734","subject":"Merge \"Drop parameters when connecting to a redirected endpoint\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/44e477dfb5821cb535d880738a965901ecd6f734"}]}],"author":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2023-11-09 14:13:49.000000000","tz":-300},"committer":{"name":"Jaromir Wysoglad","email":"jwysogla@redhat.com","date":"2023-11-09 15:37:37.000000000","tz":-300},"subject":"Fix \"dictionary changed size during iteration\"","message":"Fix \"dictionary changed size during iteration\"\n\nWe have hit an error with\n\"dictionary changed size during iteration\" a few times during\nthis week in telemetry integration tests. It seems like it\u0027s\nhitting only our ubuntu based jobs, I haven\u0027t seen this error\nin a centos based job yet. Example of the failed job can be\nfound in [1], I extracted a traceback, which leads to keystoneauth1\ninto [2]. According to [3] and [4] using copy() should help with\nthe issue. The python docs [5] indicate, that copy() should always\nbe used when iterating through sys.modules\n\n[1] https://zuul.opendev.org/t/openstack/build/c99db592871a441e9cddad2f4e60c2fc/console\n[2] https://paste.opendev.org/show/bpzng2EUyFh1tvBHczt7/\n[3] https://github.com/python/cpython/issues/84507\n[4] https://github.com/python/cpython/issues/89516\n[5] https://docs.python.org/3/library/sys.html#sys.modules\n\nChange-Id: I50500c6a21bbe60050303cea4628ca9b71a3e0eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/3b492a7aa0bcb73c7fff62c8e7b1cd5d42118fbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/3b492a7aa0bcb73c7fff62c8e7b1cd5d42118fbe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
