)]}'
{"id":"openstack%2Fkeystoneauth~909561","triplet_id":"openstack%2Fkeystoneauth~master~Ieada594dc68e152e176969258f098f7d064801f4","project":"openstack/keystoneauth","branch":"master","topic":"bug/2053965","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-07-10 13:59:28.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-07-26 15:58:35.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-07-10 14:40:50.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"36776":{"account":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"last_update":"2024-07-26 15:58:35.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-07-26 14:38:47.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-07-26 15:58:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieada594dc68e152e176969258f098f7d064801f4","subject":"Update requests of device flow for public client","status":"MERGED","created":"2024-02-20 13:20:45.000000000","updated":"2024-07-26 15:59:36.000000000","submitted":"2024-07-26 15:58:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909561-bug/2053965","meta_rev_id":"6ff234a3b0a76e73eda1fe3ddddc8bf0aa883f90","_number":909561,"virtual_id_number":909561,"owner":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-26 15:58:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-07-10 14:40:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2024-07-26 14:38:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2024-07-26 14:38:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-20 15:21:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-25 23:29:15.000000000","updated_by":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-02-25 23:29:15.000000000","updated_by":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-02-25 23:29:15.000000000","updated_by":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-07-05 14:20:07.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2024-07-08 04:37:35.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-07-10 13:59:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REMOVED"}],"messages":[{"id":"0d116a676072aa973f441df02f24b8bbbc3f8c94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"date":"2024-02-20 13:20:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fe608abbf67a5954de7b8718e9e74041dbee185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 15:21:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d25ed7ce85754ecfbb17d3320e84ebe5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d88dec71dc01492a83aadbabfda237f8 : SUCCESS in 1h 53m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d49cfe7a931450cb354b7b5abb8cc36 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8511b93d7e5b49a188548b725fe227dc : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f8823d23d784f0eb46b43cb5ed1b6dc : SUCCESS in 9m 33s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/d29397faad43407886d41b5b4d2d488e : SUCCESS in 54m 29s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/185722cd79cd4ac88b5d9236cee4a65d : SUCCESS in 4m 19s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/3cf1cce645514956a6d8caa9cceb6eba : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f0897a2129943f9a65567d46f17c998 : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"841a04b4be7c74d989d3067f318a79d6ddd177ea","author":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"date":"2024-06-26 02:40:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fc81ab2ec117d929626dd33c2531824499a40960","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-07-05 14:20:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"78de7f9dcb35d18c82e90e82b5a44c55b7bee4bd","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-07-05 20:30:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"531c5b88ec10fe81743d456d2a52642b24e83b0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 21:51:51.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/keystoneauth for 909561,1","accounts_in_message":[],"_revision_number":1},{"id":"0fd35713b6dd486a4a40848e8e2009114e6771e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"date":"2024-07-08 00:56:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"86d4afc3c8cfe02e44cfcffec82f05e8a2881fb9","author":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"date":"2024-07-08 01:01:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9abc485b137cff20524c250d46673409b6448fac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 02:37:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d7b36d09e404dbea636de7353bc3e85\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/071e54d4b4a249a78d1ece6c65457c0f : SUCCESS in 1h 35m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8359b61d9864b3b8f3be58fb9329a99 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f12a1c8f08db4f048a8433e13dc2210c : SUCCESS in 6m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d922d31e78464b689c16d2d5147d8266 : SUCCESS in 7m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/70d5d709921b43eeb94c81b8acf32242 : SUCCESS in 9m 59s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/0b62805e62f7437788360a826f91fcb5 : SUCCESS in 1h 05m 01s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/93d7c032ad754291a1383565bf7102e0 : SUCCESS in 3m 37s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/63afc26aa36b47ee94b8377bf2660d68 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/400c8df2ba3a465eac57df09022fc1c3 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"565fbc8e3ce29915415ff53c77347e2055d25895","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-08 04:37:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b3fa9ffb6b2ef7470784b60b2df7b6269b8d8d18","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-07-08 13:01:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8cd0d428fcac7cc2b6d8a09b61482d5e1321c720","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"date":"2024-07-09 00:31:22.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"81740a1f969c44e10f30c7da43f8ba23ff8325e6","author":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"date":"2024-07-09 00:50:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"69da45d868784d7b1ea4378b5aad68bcfcd69267","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 02:17:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/489848ffaa264b17ba41eae84a972f94\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bbaa06add51449b2bbce0c8a60010120 : SUCCESS in 1h 40m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c09e95ba7b84453288885286733c4a87 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca114b1d5731473b907c6e010ecaf4aa : SUCCESS in 6m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13a50c32d24b4d1d88bc76ace2b9753b : SUCCESS in 6m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dc30dd5ede0b431595034c55d7690fb1 : SUCCESS in 10m 58s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/f7e7a609a3c8443f8ae250e82193dc63 : SUCCESS in 58m 02s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/f55335a829364b858150437c9df329c7 : SUCCESS in 4m 14s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/9dd0d998777140028d2e6c7049ed794e : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01cada37c1c5431295b843430bc50870 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4efce5a2b1ad4f478954323a91256109 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":3},{"id":"f2781dae3f36e1dbd9fe0c5828d812bf30b499bd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-10 13:59:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5b74eee5dfc2ebdc157ccbb53cf07d2a1bc4c89d","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-07-10 14:40:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3aceb682e512aa6e6276bc43fa24feb2e3ea994e","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-07-26 14:38:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"18fa23146b977ab6a1b0c175e8144a82d0788d5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 14:39:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"46ad9aaa096a92bed5e734ed36079f3314b3ddb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 15:58:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0deb38a507b4eb49278f63777ac0a9c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74235ab584394d32ae13c21d5ace0e38 : SUCCESS in 1h 18m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/631fa9b0ba7a4298897414c5adfb3379 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0b87645cb1b49408eeb5cad66dcd62e : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61e4f0e69537436185bdbecde19490c9 : SUCCESS in 5m 44s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/48fdf6dc679544b6a4fcb8ef1020901b : SUCCESS in 57m 46s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/82356a07acd54d6a9423fa6cf1fec1d0 : SUCCESS in 4m 24s\n- osc-tox-py310-tips https://zuul.opendev.org/t/openstack/build/52c754cb3538402f8fefe91a7d9eec54 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15ea3a80669248039a89f76043648e23 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d29922e04bea4e3297c1300a62dbf4fb : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":3},{"id":"f50fa910c48e80504b5380da12d4ad8b5680d024","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 15:58:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6ff234a3b0a76e73eda1fe3ddddc8bf0aa883f90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 15:59:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fa612e2727741d8bef4f5ad87b32860\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3693ca84722c4222991454cd1968efb1 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8beac917983443a1a9c9f69a056da4d0 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0b82d4283a5f1424b0c3f6a6ec597e4c80e36f99","revisions":{"079218cdb3d8c6b9e174d228ddfcc7f78285eb53":{"kind":"REWORK","_number":1,"created":"2024-02-20 13:20:45.000000000","uploader":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"ref":"refs/changes/61/909561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/61/909561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/61/909561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/61/909561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/61/909561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/61/909561/1"}}},"commit":{"parents":[{"commit":"e071ad44b01362b8ee930cb23dbb52e2d87fd62d","subject":"reno: Update master for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/e071ad44b01362b8ee930cb23dbb52e2d87fd62d"}]}],"author":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@linecorp.com","date":"2024-02-05 00:45:31.000000000","tz":540},"committer":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2024-02-20 07:42:43.000000000","tz":540},"subject":"Update requests of device flow for public client","message":"Update requests of device flow for public client\n\nThis commit makes a change in OIDC Device Flow to include the client_id\nin the request body without performing basic authentication when no\nclient_secret is specified.\n\nAccording to RFC8628, in the Device Authorization Request [1] and Device\nAccess Token Request [2], if the client is not authenticating with the\nauthorization server, the client_id should be included in the request\nbody.\n\nIn addition, this commit adds missing unit tests for the\nOidcDeviceAuthorization.\n\n[1] https://datatracker.ietf.org/doc/html/rfc8628#section-3.1\n[2] https://datatracker.ietf.org/doc/html/rfc8628#section-3.4\n\nCloses-bug: #2053965\nChange-Id: Ieada594dc68e152e176969258f098f7d064801f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/079218cdb3d8c6b9e174d228ddfcc7f78285eb53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/079218cdb3d8c6b9e174d228ddfcc7f78285eb53"}]},"branch":"refs/heads/master"},"993231af03a27d1d3b49b4299c5f3e13ad4eafa6":{"kind":"REWORK","_number":2,"created":"2024-07-08 00:56:08.000000000","uploader":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"ref":"refs/changes/61/909561/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/61/909561/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/61/909561/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/61/909561/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/61/909561/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/61/909561/2"}}},"commit":{"parents":[{"commit":"d552a9a1b7a00d753b83866f9b551fe177de649c","subject":"Add OTP to v3OIDCpassword plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/d552a9a1b7a00d753b83866f9b551fe177de649c"}]}],"author":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2024-07-08 00:54:52.000000000","tz":540},"committer":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2024-07-08 00:55:38.000000000","tz":540},"subject":"Update requests of device flow for public client","message":"Update requests of device flow for public client\n\nThis commit makes a change in OIDC Device Flow to include the client_id\nin the request body without performing basic authentication when no\nclient_secret is specified.\n\nAccording to RFC8628, in the Device Authorization Request [1] and Device\nAccess Token Request [2], if the client is not authenticating with the\nauthorization server, the client_id should be included in the request\nbody.\n\nIn addition, this commit adds missing unit tests for the\nOidcDeviceAuthorization.\n\n[1] https://datatracker.ietf.org/doc/html/rfc8628#section-3.1\n[2] https://datatracker.ietf.org/doc/html/rfc8628#section-3.4\n\nCloses-bug: #2053965\nChange-Id: Ieada594dc68e152e176969258f098f7d064801f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/993231af03a27d1d3b49b4299c5f3e13ad4eafa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/993231af03a27d1d3b49b4299c5f3e13ad4eafa6"}]},"branch":"refs/heads/master"},"0b82d4283a5f1424b0c3f6a6ec597e4c80e36f99":{"kind":"REWORK","_number":3,"created":"2024-07-09 00:31:22.000000000","uploader":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"ref":"refs/changes/61/909561/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/61/909561/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/61/909561/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/61/909561/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/61/909561/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/61/909561/3"}}},"commit":{"parents":[{"commit":"d552a9a1b7a00d753b83866f9b551fe177de649c","subject":"Add OTP to v3OIDCpassword plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/d552a9a1b7a00d753b83866f9b551fe177de649c"}]}],"author":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2024-07-08 00:54:52.000000000","tz":540},"committer":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2024-07-09 00:30:55.000000000","tz":540},"subject":"Update requests of device flow for public client","message":"Update requests of device flow for public client\n\nThis commit makes a change in OIDC Device Flow to include the client_id\nin the request body without performing basic authentication when no\nclient_secret is specified.\n\nAccording to RFC8628, in the Device Authorization Request [1] and Device\nAccess Token Request [2], if the client is not authenticating with the\nauthorization server, the client_id should be included in the request\nbody.\n\nIn addition, this commit adds missing unit tests for the\nOidcDeviceAuthorization.\n\n[1] https://datatracker.ietf.org/doc/html/rfc8628#section-3.1\n[2] https://datatracker.ietf.org/doc/html/rfc8628#section-3.4\n\nCloses-bug: #2053965\nChange-Id: Ieada594dc68e152e176969258f098f7d064801f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/0b82d4283a5f1424b0c3f6a6ec597e4c80e36f99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/0b82d4283a5f1424b0c3f6a6ec597e4c80e36f99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
