)]}'
{"id":"openstack%2Fkeystoneauth~928807","triplet_id":"openstack%2Fkeystoneauth~master~I91dee65bd545fcc7b01f423253eab3f7dfc6cda3","project":"openstack/keystoneauth","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-09-27 16:17:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I91dee65bd545fcc7b01f423253eab3f7dfc6cda3","subject":"typing: Make abstract method actually abstract","status":"MERGED","created":"2024-09-10 14:06:30.000000000","updated":"2024-09-27 16:18:12.000000000","submitted":"2024-09-27 16:17:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928807-typing","meta_rev_id":"0fdad7c95427fcf1d55818ecd47aa9b5e9426716","_number":928807,"virtual_id_number":928807,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-27 16:17:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-09-27 14:31:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-11 17:44:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2024-09-27 14:31:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-10 16:03:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-11 15:37:18.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2024-09-27 14:31:06.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"bc3869b2b17cd539eabce16e87adae046e140f3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-10 14:06:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c335b221b43336d90ff2f1d2dd4304f9e2b3f81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 16:03:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cafe02d362cb4443ab5cc5aef3d16ccf\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4cdaf6a79aec40e7a9236a24a0d577ff : SUCCESS in 1h 49m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/654cd45cf3c844a3b6b6e80312deb837 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/beef31f470f14614a46ca870c048eeb8 : SUCCESS in 7m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25edf315456c42a884908fe3d7539cfd : SUCCESS in 7m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e1d77d30f4334564967560047f13f03d : SUCCESS in 9m 06s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/c0d842553a204ff4a0dc0ef4963e9744 : SUCCESS in 1h 02m 57s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/e503f542e16f4f57b55d5c9a1c195754 : SUCCESS in 4m 06s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/cc68921e83234011bbb81a2a97eef3b5 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b24542922fc42fb99ac43e5244935d2 : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"66824104d79073eeaf62de1c5e738579bf531260","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-09-11 15:37:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f62817d4027906574c9ef35fd7bed4e80836f317","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-11 17:44:56.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"eaa9f838cefb4fbea1e623375b6879974c926c57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 19:32:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcd0b7c2b009444d91c271646645e5be\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79c4b75dfb474a398befcf907d626772 : SUCCESS in 1h 46m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6ad354a9b564003a992445ebe600266 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5074240d5da34690899c0a59059d1994 : SUCCESS in 8m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11b96f38de6e4f98a9a211d8f622c21d : SUCCESS in 7m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08761d47c31649a08401d77640c6ff9f : SUCCESS in 8m 50s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/6ce8bfd218254c8ea4f698cfb84d3fe4 : SUCCESS in 52m 52s\n- osc-tox-py38-tips https://zuul.opendev.org/t/openstack/build/43446f13398941db9ee1af1b78b5f73d : SUCCESS in 5m 32s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/91faba5431984d0ab637202e7560e637 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f95a6d4ba9c45839ecf46d452be3f9d : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":2},{"id":"83a2c7d6ec1c93ab5e8e86e125d11652145699e7","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-09-27 14:31:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6884a401f0e707c39aae6a7ea829694fef60eea1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 14:31:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6a10182dce2140986e47639171c26485d622a0fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 16:17:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8ea1d49313d4a0c964b4737887c1f92\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c8a9dcc0b834c908c24d6ec27233913 : SUCCESS in 1h 44m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a1df9dd12b44e37900ecafad5b1f268 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05e6e2ffe5334231bd0f548aa5e39d9f : SUCCESS in 8m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ab57b38978f4313a14e7959198b37d7 : SUCCESS in 7m 21s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8bbe72f5bf114408b3d92ef16d6fe222 : SUCCESS in 18m 11s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f0ab3ac0213f4cff94044ee91eade9b4 : SUCCESS in 3m 08s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/9f3e4dc057b74179b4f57d23dd7bd2bc : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb5bb89e84d54492a982774c5b9fd982 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"e5269c2e743fa61410e5a85fcb87a1d4e6ef4f99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 16:17:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0fdad7c95427fcf1d55818ecd47aa9b5e9426716","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 16:18:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7be0ca2d054d4c978deb9238ffe951d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef11e838fa8e4c61b9324462aa9d4acf : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f07049bce985d1e2f8465e626d585b8b30cce2ef","revisions":{"a8e1a95b2da4b162fa15c8fa9f5d6bd25f9d0112":{"kind":"REWORK","_number":1,"created":"2024-09-10 14:06:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/07/928807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/07/928807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/07/928807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/07/928807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/07/928807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/07/928807/1"}}},"commit":{"parents":[{"commit":"ea67c1bcf5717cc85a39beda3fa3b484b3bfeec3","subject":"Enable ruff, ruff-format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/ea67c1bcf5717cc85a39beda3fa3b484b3bfeec3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-08-02 10:56:57.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-10 09:41:58.000000000","tz":60},"subject":"typing: Make abstract method actually abstract","message":"typing: Make abstract method actually abstract\n\nThis was a weird way to achieve this and mypy isn\u0027t happy since the\ndefinition is not the same as the parent. Fix it.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I91dee65bd545fcc7b01f423253eab3f7dfc6cda3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/a8e1a95b2da4b162fa15c8fa9f5d6bd25f9d0112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/a8e1a95b2da4b162fa15c8fa9f5d6bd25f9d0112"}]},"branch":"refs/heads/master"},"f07049bce985d1e2f8465e626d585b8b30cce2ef":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-09-11 17:44:56.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/07/928807/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/07/928807/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/07/928807/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/07/928807/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/07/928807/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/07/928807/2"}}},"commit":{"parents":[{"commit":"2c6f13e1f1d3f4d3cc176edd2f59245cbc2089cd","subject":"Enable ruff, ruff-format","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/2c6f13e1f1d3f4d3cc176edd2f59245cbc2089cd"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-08-02 10:56:57.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-09-11 17:43:59.000000000","tz":60},"subject":"typing: Make abstract method actually abstract","message":"typing: Make abstract method actually abstract\n\nThis was a weird way to achieve this and mypy isn\u0027t happy since the\ndefinition is not the same as the parent. Fix it.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I91dee65bd545fcc7b01f423253eab3f7dfc6cda3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/f07049bce985d1e2f8465e626d585b8b30cce2ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/f07049bce985d1e2f8465e626d585b8b30cce2ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
