)]}'
{"id":"openstack%2Fkeystoneauth~947208","triplet_id":"openstack%2Fkeystoneauth~master~I6e44e70394d51c9fca06d7f511d1ffe0d9b2c025","project":"openstack/keystoneauth","branch":"master","topic":"bug/2107373","attention_set":{},"removed_from_attention_set":{"36776":{"account":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"last_update":"2025-04-18 16:26:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6e44e70394d51c9fca06d7f511d1ffe0d9b2c025","subject":"Fix bug in OIDC Device Authorization","status":"MERGED","created":"2025-04-15 10:09:11.000000000","updated":"2025-04-18 16:27:39.000000000","submitted":"2025-04-18 16:26:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947208-bug/2107373","meta_rev_id":"cdc780e7073023cfa50d56198ed4bf761ad8305b","_number":947208,"virtual_id_number":947208,"owner":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-18 16:26:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2025-04-18 14:43:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2025-04-16 18:44:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2025-04-18 14:46:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-15 11:24:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-16 18:44:16.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2025-04-18 14:43:15.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"e4e22a7fda8971e686e6c6288f358669685ec747","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"date":"2025-04-15 10:09:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca81cad10c80c5edd1842888ea12b54abe0cdecd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"date":"2025-04-15 10:10:30.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c7c30893e4c243cd678c0d7e7d8210777fefa251","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"date":"2025-04-15 10:25:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"09ab02b80fe5ce5f5439b2d44e8b35dbf416cbda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 11:24:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa1ad89b685647e8a6d2d76963804cfa\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c89165474ef45c18286a811814bb9f1 : SUCCESS in 51m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7f9d2937fd34c14b2a942ea25966760 : SUCCESS in 1m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7086975350842d685745d8a00416fc6 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c045543b1c554f40af7fbb89dd907502 : SUCCESS in 7m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/34f838c51a064658907ad207928b26bd : SUCCESS in 12m 08s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/9d41fd6f6b5e43eba0c61cf5a7543ba6 : SUCCESS in 58m 45s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3afcd121168445adad83f1b423fa652a : SUCCESS in 4m 17s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/c59b1bc0a91e4fcfaec1c9afebad3d78 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa1d1bb4c0534d5ebb93fd066a2d147c : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":3},{"id":"27b88ba582dc313f02e50a312b7a9e02cbc8fc37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"date":"2025-04-15 11:48:05.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"add7a0874a31966cdae0197523d35bd6bc6564b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 13:24:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/553e59004bb64841993f205b35d8c0eb\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3ef4ff49f4c4a7692d9e160ab50f129 : SUCCESS in 1h 31m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a9b4d99d0ec4200befe0368cab7b570 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d2c3e24f2224acc84a41f455fc31538 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0087bc3beaed420bb875feeff3c787aa : SUCCESS in 9m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45e5b88799704b86aab5e268a6976005 : SUCCESS in 10m 36s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/ec9c2b8570fc411b94783c7cd4c5c0cd : SUCCESS in 52m 13s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/571f5a597f4e4f38889763321574b204 : SUCCESS in 6m 35s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/4feef47b780a4ace907dc9e4f4d17886 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59b3e52b25434cde98f603b391dd9de3 : SUCCESS in 9m 38s","accounts_in_message":[],"_revision_number":4},{"id":"1bd1a1ad91f3adaffc71173701ce352037b3cf5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"date":"2025-04-15 13:30:07.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"2e8f98a9d83a5adb044d841ee0ff6f8e9c9d95e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 15:07:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44236e33bcdd4c79a0dd7fb697650e85\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6f7af9674964e8982dfbb8345966aea : SUCCESS in 1h 33m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df678e8ce65542f198910997ad9841f0 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/961375f7b47c411291d1263caf67b571 : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0d5f8fdd33f64778bfda35dfceee20cc : SUCCESS in 7m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4e1212787be41958b4e295f00f3b39d : SUCCESS in 10m 22s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/052f8a1352f949ec99caed0b37cbe8d6 : SUCCESS in 58m 18s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6aa0bea5e5aa4122b365bfad64c437ae : SUCCESS in 5m 15s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/65819fe0f9a342a68ab2f9cad30bac63 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89f0004c8fd7482498270184dcb83b6c : SUCCESS in 8m 04s","accounts_in_message":[],"_revision_number":5},{"id":"2a3e0fb7d0c7f3c96fa69dfb3d237d63fe16af6c","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-04-16 18:44:16.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"58e23947f83408183687c951fb0d920984614b0f","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2025-04-18 14:43:15.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"18af6f204a2cdaf441a923491c63de94f28bbe65","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-04-18 14:46:09.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f814e6b9352821a02fb023d6d236edc70016d2b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 14:46:35.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0f786d903fd887c18925237a4c6cb5e9b949506b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 16:26:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afb10052f3be4b4791ed652d760c10a7\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/764951322eb1448b8183409cd37f283e : SUCCESS in 1h 40m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7419976ca934f64b36c4f9adbaace63 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1884067edc394637baa07653863941e3 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44384b48dca446dea5325a64bfff9dbb : SUCCESS in 6m 39s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/55124025bec94cc98bdb22a8c633e720 : SUCCESS in 1h 02m 45s\n- osc-tox-py39-tips https://zuul.opendev.org/t/openstack/build/26145f666a9c47f3b9216d350684b25a : SUCCESS in 4m 37s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/eabed5a367f94ea78d029643362bd0a1 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6efbe428ed3486d89eaabba8ec343e9 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":5},{"id":"b0b03e684081e88559e1b38bbcfe23d31339ee3a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 16:26:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"cdc780e7073023cfa50d56198ed4bf761ad8305b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 16:27:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7676a5a449984455ac499f4e6bc77ecf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eaa003060064512be2204c71a9bf672 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"884578fd001a88bb8996c599095f748f7c3f1d13","revisions":{"eec5ec6526320076d9560703392b5fd8cdcdcec2":{"kind":"REWORK","_number":1,"created":"2025-04-15 10:09:11.000000000","uploader":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"ref":"refs/changes/08/947208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/08/947208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/1"}}},"commit":{"parents":[{"commit":"43bc36dd59c34db64ab3305ba530ec16e0486c12","subject":"Merge \"Use releases.openstack.org instead of opendev.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/43bc36dd59c34db64ab3305ba530ec16e0486c12"}]}],"author":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2025-04-15 09:24:34.000000000","tz":540},"committer":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2025-04-15 10:09:05.000000000","tz":540},"subject":"Fix bug in OIDC Device Authorization","message":"Fix bug in OIDC Device Authorization\n\nThere is two bugs in the current device flow using keystoneauth:\n\n1. An issue where an error occurs if `client_secret` is not specified.\nIn the device flow, authentication should be possible without specifying\n`client_secret`[1] but the current implementation requires it.\nThis bug was introduced by the following patch:\n- https://review.opendev.org/c/openstack/keystoneauth/+/932989\n\n2. An issue with the polling process from the client in the device flow\nIn the device flow, polling is used to verify that the user has\nauthenticated. There is an issue with the process that checks for\nincomplete polling, causing the process to halt with an error:\n\"Unrecognized schema in response body. (HTTP 400)\"\nThis bug was introduced by the following patch update:\nhttps://review.opendev.org/c/openstack/keystoneauth/+/930769\n\n[1] https://datatracker.ietf.org/doc/html/rfc8628#section-5.6\n\nCloses-bug: #2107373\nChange-Id: I6e44e70394d51c9fca06d7f511d1ffe0d9b2c025\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/eec5ec6526320076d9560703392b5fd8cdcdcec2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/eec5ec6526320076d9560703392b5fd8cdcdcec2"}]},"branch":"refs/heads/master"},"91f2d88883573071730a14eb4ed93b2a1d0098fd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-04-15 10:10:30.000000000","uploader":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"ref":"refs/changes/08/947208/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/08/947208/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/2"}}},"commit":{"parents":[{"commit":"43bc36dd59c34db64ab3305ba530ec16e0486c12","subject":"Merge \"Use releases.openstack.org instead of opendev.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/43bc36dd59c34db64ab3305ba530ec16e0486c12"}]}],"author":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2025-04-15 09:24:34.000000000","tz":540},"committer":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2025-04-15 10:10:13.000000000","tz":540},"subject":"Fix bug in OIDC Device Authorization","message":"Fix bug in OIDC Device Authorization\n\nThere are two bugs in the current device flow using keystoneauth:\n\n1. An issue where an error occurs if `client_secret` is not specified.\nIn the device flow, authentication should be possible without specifying\n`client_secret`[1] but the current implementation requires it.\nThis bug was introduced by the following patch:\n- https://review.opendev.org/c/openstack/keystoneauth/+/932989\n\n2. An issue with the polling process from the client in the device flow\nIn the device flow, polling is used to verify that the user has\nauthenticated. There is an issue with the process that checks for\nincomplete polling, causing the process to halt with an error:\n\"Unrecognized schema in response body. (HTTP 400)\"\nThis bug was introduced by the following patch update:\nhttps://review.opendev.org/c/openstack/keystoneauth/+/930769\n\n[1] https://datatracker.ietf.org/doc/html/rfc8628#section-5.6\n\nCloses-bug: #2107373\nChange-Id: I6e44e70394d51c9fca06d7f511d1ffe0d9b2c025\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/91f2d88883573071730a14eb4ed93b2a1d0098fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/91f2d88883573071730a14eb4ed93b2a1d0098fd"}]},"branch":"refs/heads/master"},"ac482080cead27005ca7b79d1c28eeb107078e5c":{"kind":"REWORK","_number":3,"created":"2025-04-15 10:25:09.000000000","uploader":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"ref":"refs/changes/08/947208/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/08/947208/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/3"}}},"commit":{"parents":[{"commit":"43bc36dd59c34db64ab3305ba530ec16e0486c12","subject":"Merge \"Use releases.openstack.org instead of opendev.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/43bc36dd59c34db64ab3305ba530ec16e0486c12"}]}],"author":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2025-04-15 09:24:34.000000000","tz":540},"committer":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2025-04-15 10:24:50.000000000","tz":540},"subject":"Fix bug in OIDC Device Authorization","message":"Fix bug in OIDC Device Authorization\n\nThere are two bugs in the current device flow using keystoneauth:\n\n1. An issue where an error occurs if `client_secret` is not specified.\nIn the device flow, authentication should be possible without specifying\n`client_secret`[1] but the current implementation requires it.\nThis bug was introduced by the following patch:\n- https://review.opendev.org/c/openstack/keystoneauth/+/932989\n\n2. An issue with the polling process from the client in the device flow\nIn the device flow, polling is used to verify that the user has\nauthenticated. There is an issue with the process that checks for\nincomplete polling, causing the process to halt with an error:\n\"Unrecognized schema in response body. (HTTP 400)\"\nThis bug was introduced by the following patch update:\nhttps://review.opendev.org/c/openstack/keystoneauth/+/930769\n\n[1] https://datatracker.ietf.org/doc/html/rfc8628#section-5.6\n\nCloses-bug: #2107373\nChange-Id: I6e44e70394d51c9fca06d7f511d1ffe0d9b2c025\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/ac482080cead27005ca7b79d1c28eeb107078e5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/ac482080cead27005ca7b79d1c28eeb107078e5c"}]},"branch":"refs/heads/master"},"4e601c6c13e7a0dc4f46978171bfd0978931acbd":{"kind":"REWORK","_number":4,"created":"2025-04-15 11:48:05.000000000","uploader":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"ref":"refs/changes/08/947208/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/08/947208/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/4"}}},"commit":{"parents":[{"commit":"43bc36dd59c34db64ab3305ba530ec16e0486c12","subject":"Merge \"Use releases.openstack.org instead of opendev.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/43bc36dd59c34db64ab3305ba530ec16e0486c12"}]}],"author":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2025-04-15 09:24:34.000000000","tz":540},"committer":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2025-04-15 11:42:30.000000000","tz":540},"subject":"Fix bug in OIDC Device Authorization","message":"Fix bug in OIDC Device Authorization\n\nThere are two bugs in the current device flow using keystoneauth:\n\n1. An issue where an error occurs if `client_secret` is not specified.\nIn the device flow, authentication should be possible without specifying\n`client_secret`[1] but the current implementation requires it.\nThis bug was introduced by the following patch:\n- https://review.opendev.org/c/openstack/keystoneauth/+/932989\n\n2. An issue with the polling process from the client in the device flow\nIn the device flow, polling is used to verify that the user has\nauthenticated. There is an issue with the process that checks for\nincomplete polling, causing the process to halt with an error:\n\"Unrecognized schema in response body. (HTTP 400)\"\nThis bug was introduced by the following patch update:\nhttps://review.opendev.org/c/openstack/keystoneauth/+/930769\n\n[1] https://datatracker.ietf.org/doc/html/rfc8628#section-5.6\n\nCloses-bug: #2107373\nChange-Id: I6e44e70394d51c9fca06d7f511d1ffe0d9b2c025\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/4e601c6c13e7a0dc4f46978171bfd0978931acbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/4e601c6c13e7a0dc4f46978171bfd0978931acbd"}]},"branch":"refs/heads/master"},"884578fd001a88bb8996c599095f748f7c3f1d13":{"kind":"REWORK","_number":5,"created":"2025-04-15 13:30:07.000000000","uploader":{"_account_id":36776,"name":"Tatsuya Hayashino","display_name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","username":"tatsuya.hayashino"},"ref":"refs/changes/08/947208/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/08/947208/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/08/947208/5"}}},"commit":{"parents":[{"commit":"43bc36dd59c34db64ab3305ba530ec16e0486c12","subject":"Merge \"Use releases.openstack.org instead of opendev.org\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/43bc36dd59c34db64ab3305ba530ec16e0486c12"}]}],"author":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2025-04-15 09:24:34.000000000","tz":540},"committer":{"name":"Tatsuya Hayashino","email":"tatsuya.hayashino@lycorp.co.jp","date":"2025-04-15 13:28:41.000000000","tz":540},"subject":"Fix bug in OIDC Device Authorization","message":"Fix bug in OIDC Device Authorization\n\nThere are two bugs in the current device flow using keystoneauth:\n\n1. An issue where an error occurs if `client_secret` is not specified.\nIn the device flow, authentication should be possible without specifying\n`client_secret`[1] but the current implementation requires it.\nThis bug was introduced by the following patch:\n- https://review.opendev.org/c/openstack/keystoneauth/+/932989\n\n2. An issue with the polling process from the client in the device flow\nIn the device flow, polling is used to verify that the user has\nauthenticated. There is an issue with the process that checks for\nincomplete polling, causing the process to halt with an error:\n\"Unrecognized schema in response body. (HTTP 400)\"\nThis bug was introduced by the following patch update:\nhttps://review.opendev.org/c/openstack/keystoneauth/+/930769\n\n[1] https://datatracker.ietf.org/doc/html/rfc8628#section-5.6\n\nCloses-bug: #2107373\nChange-Id: I6e44e70394d51c9fca06d7f511d1ffe0d9b2c025\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/884578fd001a88bb8996c599095f748f7c3f1d13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/884578fd001a88bb8996c599095f748f7c3f1d13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
