)]}'
{"id":"openstack%2Fkeystoneauth~994713","triplet_id":"openstack%2Fkeystoneauth~master~I11899db50f4cab843dd1aff0bee9d1f0fb1a7fa6","project":"openstack/keystoneauth","branch":"master","topic":"typing","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-14 19:19:07.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I11899db50f4cab843dd1aff0bee9d1f0fb1a7fa6","subject":"typing: Use objects from typing","status":"MERGED","created":"2026-06-24 12:55:20.000000000","updated":"2026-09-14 22:32:30.000000000","submitted":"2026-09-14 22:31:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994713-typing","meta_rev_id":"bd57aaff2757dcf19cd5b18ef1fed51da810bed0","_number":994713,"virtual_id_number":994713,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-14 22:31:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"recommended":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-09-14 19:19:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-14 19:19:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2026-09-14 19:19:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-24 14:56:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-03 17:26:25.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2026-09-04 10:34:39.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"ad167f85a0c6d5c1d7aa90fb83e39dcf151232fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-24 12:55:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fd80071932800755cf91e4aed72791a8551004f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 14:56:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82685daf5e3041e79dd7ce5892515502\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e36c4862a9847a68468c5e7d7ded75d : SUCCESS in 1h 52m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0be135e32c54c4b99deba5b2b672757 : SUCCESS in 7m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/177a862c13ce4eea911ffafec701e315 : SUCCESS in 4m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5de13aae31c74e8fb5fc22b4f0f3f639 : SUCCESS in 8m 39s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c065cca7ec6e4a098a2602ab585ba110 : SUCCESS in 10m 40s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/2dd186bd818f4c03941a5e7c6f7b351f : SUCCESS in 15m 29s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/df22492c59c842c281abb6038246fbc4 : SUCCESS in 1h 13m 31s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8f51d6ef304748f48f842a987f8de6a4 : SUCCESS in 2m 47s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/f29aef4264384e1b96f04d4d8430e214 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/754f210fbb3d4f6881963026b0e9872e : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3aaeb0d10f343fca6b696c0f774b085 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"c257a3ec9b74b2d102f00691af126efa2ce50caf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-27 00:43:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b2e46081b796a16f94e7e588e318db104d43cbea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-27 00:44:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"064ad9444d050dbff49cd5f2ed382eeaa6a1d5aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-27 02:40:23.000000000","message":"Patch Set 1: Verified-2\n\n(7 comments)\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3ab1be28eb9420e9c118ccf99817451\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/06191092f95c46bbb79df603a35f5a29 : SUCCESS in 54m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ed5b1c7b4134e3fb392352ff2b184b8 : FAILURE in 7m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/041065b9ddf54714b0aec2d9e52a41c2 : SUCCESS in 7m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cd924ed9d569488181217182b78dfb1f : SUCCESS in 6m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7c08784ae7bb417596cf2bb0521b0902 : SUCCESS in 9m 44s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/34c5ada4cda44426a38a5f8e56a58a19 : SUCCESS in 1h 04m 02s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/79e1d332b8d84f499d5d08fc5732edc3 : SUCCESS in 4m 23s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/67627651bc694f59acb1a3cea04131da : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b5830811f294c89a9c1a8ba6fa2d504 : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc39853bb6104a21bc99b11d72724aff : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d700115256904f97c6c718ab70a832cb094ad97c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-12 11:26:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e1780f4ad6e4e2e049c24dfddb6f8bc684ee211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-12 12:38:01.000000000","message":"Patch Set 1: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c37fb1b824349609deb3a66aa3095b6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed35fd2073174d0fb689399f04c82e79 : SUCCESS in 1h 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5846fad573db4cb4b1ec883d11f2514a : FAILURE in 7m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/baf7f3606b894f91b397e4707ca41e3f : SUCCESS in 7m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/60c0835700ef44cbabc4b3900df23120 : SUCCESS in 6m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8681471806fa4075acf93116c11f505b : SUCCESS in 6m 09s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/bea5dde19fce44a181e40bafcc10a113 : SUCCESS in 11m 14s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/b472fcc869fd4753ad7c9a38edc121a3 : SUCCESS in 1h 01m 26s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/839c4449e60a4574b587947403fa64cd : SUCCESS in 6m 46s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/dcc9d196532949cab68a32cd7d54e177 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bb5c3f5befc4a1d82dc546267c7be0a : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87850b1448d0486fbfaa131c6fec13e7 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"98ba78bc3598f9748672b365103388d1255c3f81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-19 16:10:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"188b31465d4bfb29628314c85ae4d31f4bcbf154","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-19 17:27:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e58712dc2ecb4edbb9ff732052bf7a63\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/097fe10e3f1e437eb668b3c285343267 : SUCCESS in 1h 08m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94477f8a115545c79f9fb611d63eb028 : SUCCESS in 8m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4b758b626a5433cb5981077d9dfc7f4 : SUCCESS in 6m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b0d5e6e9b4042a7b3fed2e63d51025e : SUCCESS in 6m 12s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4684d459ceae4c6caaa810a4aaf89aab : SUCCESS in 5m 22s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/01579118fd3c48118a66c91584f93a2f : SUCCESS in 14m 52s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/2f1a91e4f09c431baa7096f210c4581c : SUCCESS in 56m 38s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/be193ec490114b0991569b5c77066d98 : SUCCESS in 4m 31s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/34dae9d7c1054e53a1b7018895e0a0f8 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f35c3c30a9e4bec81141a9dbfadef75 : SUCCESS in 9m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d7585cf4001405ba99264c6a0888604 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":2},{"id":"2d787757e81e7379f96f654d86e796fff28ab33a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-08-20 12:00:02.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"86c9fd7f6e54fd9e2ccf9827c990d4f67ecba673","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-03 13:48:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"5c7271673be55f634ff10395caee84d88f8a3483","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:31:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf0463768aaa436cb8a9a75abfefca5b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/218b566373714903ae4c576cc561f57b : SUCCESS in 43m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/584e201d22a3460aadc5816d3fa35be0 : SUCCESS in 9m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8585e6bdee040e394fb5aa9398e67ef : SUCCESS in 8m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb12a4aeafc84c4186b0d46adbfcb762 : SUCCESS in 7m 14s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/10af552ebd414b9fb37ade5ed18e654d : SUCCESS in 10m 43s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/579dd9e238b2418082459bd314eae75a : SUCCESS in 15m 12s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/8e1a165409df45aa83db023f7f1da934 : SUCCESS in 1h 17m 37s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5bf6ab823c2c48dca6d276846396d7e4 : SUCCESS in 5m 50s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/dc012324d2ce44c7bc2ec7456676504c : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61f8454160a04c628604c7c1a7161740 : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b42fffb2fb704f1bb4c3171ec3ee9268 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":3},{"id":"734ca7cd6ad5705c1541ff86cdead5cea8fafa16","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-09-03 17:26:25.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d0c64d66eb56cbd1e30a8402017ea740a6e71f1f","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-09-04 10:34:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"047e2f4dba0edda507d290efb1563a8929df36a5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-14 09:55:51.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a4faec796e792fe6b5719ab303c48957405f7089","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-14 19:19:04.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"0150de47efb3ee0bcdd0310f1b4f4045d635b3e6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-14 19:19:07.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ebff169c4626a1f4e0e0ce3cb71a7b872a13bf86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 20:45:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccb9270a34ac49abac54bfa8f26307e6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f70fc8ca82844aebcbf3d1284481e2d : SUCCESS in 1h 23m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dba26001aadf4b329d8bc88ed72bbb4d : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f2db6488b1b43b88e1e11666badeda4 : SUCCESS in 6m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/11135bac5d4c4c9ba8375448e4d47c84 : SUCCESS in 3m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2dd1a2fac803459a80a86eb6170b1c78 : SUCCESS in 10m 34s\n- openstack-tox-py315 https://zuul.opendev.org/t/openstack/build/71e6efaf42854b08affc12b43a632e3f : SUCCESS in 8m 35s (non-voting)\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/4fbd61b836374ac1aadb9ffbeed175b4 : SUCCESS in 44m 15s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/717fb56937fa4f82929d138af55277ad : SUCCESS in 4m 26s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/a354edf7f21c4eae9c6800a93d0463e7 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0563352d64b245beb9c6e2d7d0948f54 : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92962f89fb314e908841994b904190ae : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":4},{"id":"78856dbbcfa6f11d098c6ce80832ba4e9bd311a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 20:46:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a9ef006d431ca89c35074f024132dd0ecf7838b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 22:31:13.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44a2a1b3ecbe464698fc97877ff7b38d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/176127ad83e743568f7c1f9e0f4fbecf : SUCCESS in 1h 39m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/008089d992214e0fb4afb06d13d2d45a : SUCCESS in 8m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f74d56732ff4856a8cfb77ebaa625cb : SUCCESS in 6m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e775ab90e2dd4433981477908307a45e : SUCCESS in 3m 45s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/970999ec0cc34eb0b88a517c25c57a2c : SUCCESS in 10m 20s\n- openstacksdk-functional-devstack-tips https://zuul.opendev.org/t/openstack/build/a706b662e6cc452c900f01e7baa57550 : SUCCESS in 32m 57s\n- osc-tox-py311-tips https://zuul.opendev.org/t/openstack/build/fe6a280b1f364c5a8e2434196137defb : SUCCESS in 3m 12s\n- osc-tox-py314-tips https://zuul.opendev.org/t/openstack/build/a202ce5ab68c4dc7bfdf1fc9854eefb7 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b81bd1cff8447db81b21b369200a336 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8b0d6059d4e4f79ac8be2efc36b5d5d : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":4},{"id":"1e7989c2246c44851d41c461bb057cad268cfe07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 22:31:13.000000000","message":"Change has been successfully merged\n\n3 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":4},{"id":"bd57aaff2757dcf19cd5b18ef1fed51da810bed0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 22:32:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dbb215af27a4dbaa0a60a342898d6ac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cc4dd147cbc4a06b12bcbbc0c0f9855 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9e8b776334550a651cfff4f3f1b0dba3eca90e01","revisions":{"5a8fff8f71c375f156efd1f5c8e398c3be1d5802":{"kind":"REWORK","_number":1,"created":"2026-06-24 12:55:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/994713/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/13/994713/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/13/994713/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/13/994713/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/13/994713/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/13/994713/1"}}},"commit":{"parents":[{"commit":"a11207fb53373c75296259f9d59a9e980dc4cd09","subject":"trivial: Declare attributes on _BaseAdapter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/a11207fb53373c75296259f9d59a9e980dc4cd09"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 12:40:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 12:53:32.000000000","tz":60},"subject":"typing: Use objects from typing","message":"typing: Use objects from typing\n\nThis allows tools like ruff to automatically replace imports from\ntyping_extensions with those from typing when our version of Python\nis new enough. It also results in shorter, easier to read code.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I11899db50f4cab843dd1aff0bee9d1f0fb1a7fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/5a8fff8f71c375f156efd1f5c8e398c3be1d5802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/5a8fff8f71c375f156efd1f5c8e398c3be1d5802"}]},"branch":"refs/heads/master"},"3f4959c25e420828d382cf6336a874fe3489e345":{"kind":"REWORK","_number":2,"created":"2026-08-19 16:10:11.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/994713/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/13/994713/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/13/994713/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/13/994713/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/13/994713/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/13/994713/2"}}},"commit":{"parents":[{"commit":"083d2dd499fb34a626fadf26383d01a1b4d6f95f","subject":"pre-commit: Bump versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/083d2dd499fb34a626fadf26383d01a1b4d6f95f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 12:40:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-08-19 16:10:06.000000000","tz":60},"subject":"typing: Use objects from typing","message":"typing: Use objects from typing\n\nThis allows tools like ruff to automatically replace imports from\ntyping_extensions with those from typing when our version of Python is\nnew enough. It also results in shorter, easier to read code.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I11899db50f4cab843dd1aff0bee9d1f0fb1a7fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/3f4959c25e420828d382cf6336a874fe3489e345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/3f4959c25e420828d382cf6336a874fe3489e345"}]},"branch":"refs/heads/master"},"caab681f5f6bf99c1e72aaf0d3c80a5499a68e22":{"kind":"REWORK","_number":3,"created":"2026-09-03 13:48:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/994713/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/13/994713/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/13/994713/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/13/994713/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/13/994713/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/13/994713/3"}}},"commit":{"parents":[{"commit":"2ad12864d28b460f2193d891584e2873cb9724fa","subject":"Update master for stable/2026.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/2ad12864d28b460f2193d891584e2873cb9724fa"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 12:40:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 12:40:24.000000000","tz":60},"subject":"typing: Use objects from typing","message":"typing: Use objects from typing\n\nThis allows tools like ruff to automatically replace imports from\ntyping_extensions with those from typing when our version of Python is\nnew enough. It also results in shorter, easier to read code.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I11899db50f4cab843dd1aff0bee9d1f0fb1a7fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/caab681f5f6bf99c1e72aaf0d3c80a5499a68e22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/caab681f5f6bf99c1e72aaf0d3c80a5499a68e22"}]},"branch":"refs/heads/master"},"9e8b776334550a651cfff4f3f1b0dba3eca90e01":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-09-14 19:19:04.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/13/994713/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystoneauth","ref":"refs/changes/13/994713/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/13/994713/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/13/994713/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystoneauth refs/changes/13/994713/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystoneauth refs/changes/13/994713/4"}}},"commit":{"parents":[{"commit":"7524c0e2fb9997799cce8d3b7862b6ef2f7b08f1","subject":"Fix typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/7524c0e2fb9997799cce8d3b7862b6ef2f7b08f1"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-06-24 12:40:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-14 19:19:04.000000000","tz":0},"subject":"typing: Use objects from typing","message":"typing: Use objects from typing\n\nThis allows tools like ruff to automatically replace imports from\ntyping_extensions with those from typing when our version of Python is\nnew enough. It also results in shorter, easier to read code.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I11899db50f4cab843dd1aff0bee9d1f0fb1a7fa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/9e8b776334550a651cfff4f3f1b0dba3eca90e01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystoneauth/commit/9e8b776334550a651cfff4f3f1b0dba3eca90e01"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"2ad12864d28b460f2193d891584e2873cb9724fa","ours":"caab681f5f6bf99c1e72aaf0d3c80a5499a68e22","theirs":"7524c0e2fb9997799cce8d3b7862b6ef2f7b08f1","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
