)]}'
{"id":"openstack%2Fkeystonemiddleware~1006308","triplet_id":"openstack%2Fkeystonemiddleware~master~Ifecb48c7ed761d5170703b556fd6e825be4d2834","project":"openstack/keystonemiddleware","branch":"master","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-09-18 19:21:16.000000000","reason":"A robot voted negatively on a label"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2026-10-06 10:41:30.000000000","reason":"Vote got outdated and was removed: Code-Review+2"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ifecb48c7ed761d5170703b556fd6e825be4d2834","subject":"Adopt pre-commit (redux)","status":"NEW","created":"2026-09-18 17:45:53.000000000","updated":"2026-10-06 12:27:23.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"108af572e3c9a13357838d10b48c7d08c43e83a3","_number":1006308,"virtual_id_number":1006308,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-10-06 12:27:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-18 19:21:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-10-03 07:36:31.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"fb248fabb0206e9e036f5aadb2115632401be3a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-18 17:45:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09234df25940283c17322a5f12eb382abc92d9cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-18 19:21:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a90ceb3c06d949a8a16ced0d44c9d586\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae7fea62d70142528c2110f0c3d360ed : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5b9e149edd64fdb8630d5743b398853 : FAILURE in 2m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/55898e7fd0ec49b480661eefb088f39d : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/033b15e438d446688a402b756d94849d : SUCCESS in 2m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/bce8bb1d629343deae87991978eefa7c : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef9563639ca1444d9ab2320fa669f8ad : SUCCESS in 7m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3fd4d510b0e74ff2ae02539cb9b782e3 : SUCCESS in 1h 28m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d478d8e03d6c4df6942749ee7ae5b1d2 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"97e07f5ce8040ade2e8c5c119814a712a3d71e41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 11:05:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"fd2da243f5d45cbcc9d8ff4cf05688d2254ac285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 12:27:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a083d6119a314d968fadd957e7604bcb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a941fafc7d8d4e56ad8a718d6d8f155a : SUCCESS in 3m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3119172289a48c8930346d09d29f6bb : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64729e270b2d432d90da37d0bd28fedd : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c61659a5d9c45d9b293f1378b25eaed : SUCCESS in 4m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f603f39c815f4f22a232693a92b31e2e : SUCCESS in 3m 31s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/c65d729c749c4194886453a1e34eb4af : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/667aa5fb7fe74578a3da724ac61654ae : SUCCESS in 7m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/817c41647abb4040880a44ae0891b9cb : SUCCESS in 2m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dfcb89d89138492eaea989c1d138e99a : SUCCESS in 1h 18m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c966728f0c1490abd4f6c92e3fa6517 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":2},{"id":"d771bd91790cf6454359df92a67b890b5deb8f94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-09-21 21:27:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"476c4e1b2f00f3f9f31cd0234e5d6ff71b3f5def","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-21 23:14:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a826effdfc6747f6b9434ef8a4abfab3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d492cd3b6341479b9871e22415240ac2 : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9a1f2fcb4e548bfa04631ca20d0a146 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2734f70cc0ce412b9f23876b58972031 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c508b17b746f45d1bef5eee9415f393f : SUCCESS in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/10a8e712f176474e9cf38a068d0e9d8e : SUCCESS in 5m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7153cf0619514fad8fb325b547ad5e31 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10b487a91ebe4694ba366d4f537bc22b : SUCCESS in 5m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9037c6e2ac8445b18589dfe7a9a57cd5 : SUCCESS in 2m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc0f977e81664f54ab4be226cda0f52a : SUCCESS in 1h 40m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/719290ee51304fbb8ce6b67f4232d52d : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":3},{"id":"1c23eb2d46eec91b32b987024cc69acb40b41320","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-10-03 07:36:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6fdd01b6dcbea9d91e909f3416cb890c92b0ca66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-10-06 10:41:30.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"108af572e3c9a13357838d10b48c7d08c43e83a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-10-06 12:27:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42f4938797674b23a2cd9366bdeeec88\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/af2b716538c44b64847be9fa741992b9 : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a955beedf18c4a5ea94f57e5b810da8e : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab71603b8e7c4928b77814ed785c08ff : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9953f90b0f924981b9b3bfd02c2a48dd : SUCCESS in 3m 29s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/258e36f8b8044251a308f9000d6c6f90 : SUCCESS in 8m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5595367c89c64c62a9c96c99dbeca9dc : SUCCESS in 8m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/da86619c5f9d458ab7228ee1ef00ab26 : SUCCESS in 2m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/094c0d0396a54b419a81731fb50415ab : SUCCESS in 1h 39m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67439e23886c4f9ba1a323d09fba1d2b : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"cd96c9fa7dc95f138b71874bedb37e7474ab5081","revisions":{"70e4e51ddd3c3f98c005b648de3939c1599b704a":{"kind":"REWORK","_number":1,"created":"2026-09-18 17:45:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/1006308/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/08/1006308/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/1006308/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/1006308/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/1006308/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/1006308/1"}}},"commit":{"parents":[{"commit":"7f2ff8696e3f58415f914566be63698138ac343b","subject":"Add oslo.cache backend support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/7f2ff8696e3f58415f914566be63698138ac343b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 17:44:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 17:44:29.000000000","tz":60},"subject":"Adopt pre-commit (redux)","message":"Adopt pre-commit (redux)\n\nApply ruff as done in Ife05172ff48c73853dbbfd4519c542e7cedb227d\npreviously but this time without any unrelated changes.\n\nChange-Id: Ifecb48c7ed761d5170703b556fd6e825be4d2834\nSigned-off-by: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/70e4e51ddd3c3f98c005b648de3939c1599b704a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/70e4e51ddd3c3f98c005b648de3939c1599b704a"}]},"branch":"refs/heads/master"},"d4af83111d2db3a2ec5815e652fe8defdc58db49":{"kind":"REWORK","_number":2,"created":"2026-09-21 11:05:56.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/1006308/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/08/1006308/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/1006308/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/1006308/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/1006308/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/1006308/2"}}},"commit":{"parents":[{"commit":"7f2ff8696e3f58415f914566be63698138ac343b","subject":"Add oslo.cache backend support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/7f2ff8696e3f58415f914566be63698138ac343b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 17:44:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 11:05:14.000000000","tz":60},"subject":"Adopt pre-commit (redux)","message":"Adopt pre-commit (redux)\n\nApply ruff as done in Ife05172ff48c73853dbbfd4519c542e7cedb227d\npreviously but this time without any unrelated changes. However, we make\nthe following changes:\n\n- Move ruff configuration to pyproject.toml\n- Set line-length to 79\n- Bump all pre-commit hooks to the latest version\n- Disable codes for ruff-check failures so we can address them in a\n  follow-up, making this easier to validate\n\nChange-Id: Ifecb48c7ed761d5170703b556fd6e825be4d2834\nSigned-off-by: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d4af83111d2db3a2ec5815e652fe8defdc58db49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d4af83111d2db3a2ec5815e652fe8defdc58db49"}]},"branch":"refs/heads/master"},"e632db11448f7e97dcbaa533133875bbde88f4fc":{"kind":"REWORK","_number":3,"created":"2026-09-21 21:27:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/1006308/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/08/1006308/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/1006308/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/1006308/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/1006308/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/1006308/3"}}},"commit":{"parents":[{"commit":"992f53aa6a17c82fd70352e74dffa3027f25b25e","subject":"Add oslo.cache backend support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/992f53aa6a17c82fd70352e74dffa3027f25b25e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 17:44:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-21 21:27:03.000000000","tz":60},"subject":"Adopt pre-commit (redux)","message":"Adopt pre-commit (redux)\n\nApply ruff as done in Ife05172ff48c73853dbbfd4519c542e7cedb227d\npreviously but this time without any unrelated changes. However, we make\nthe following changes:\n\n- Move ruff configuration to pyproject.toml\n- Set line-length to 79\n- Bump all pre-commit hooks to the latest version\n- Disable codes for ruff-check failures so we can address them in a\n  follow-up, making this easier to validate\n\nChange-Id: Ifecb48c7ed761d5170703b556fd6e825be4d2834\nSigned-off-by: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e632db11448f7e97dcbaa533133875bbde88f4fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/e632db11448f7e97dcbaa533133875bbde88f4fc"}]},"branch":"refs/heads/master"},"cd96c9fa7dc95f138b71874bedb37e7474ab5081":{"kind":"REWORK","_number":4,"created":"2026-10-06 10:41:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/1006308/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/08/1006308/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/1006308/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/1006308/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/1006308/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/08/1006308/4"}}},"commit":{"parents":[{"commit":"9e018d8fa6bdfe066129aeaf95a752134ce2b182","subject":"Add oslo.cache backend support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/9e018d8fa6bdfe066129aeaf95a752134ce2b182"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-09-18 17:44:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2026-10-06 10:39:35.000000000","tz":60},"subject":"Adopt pre-commit (redux)","message":"Adopt pre-commit (redux)\n\nApply ruff as done in Ife05172ff48c73853dbbfd4519c542e7cedb227d\npreviously but this time without any unrelated changes. However, we make\nthe following changes:\n\n- Move ruff configuration to pyproject.toml\n- Set line-length to 79\n- Bump all pre-commit hooks to the latest version\n- Disable codes for ruff-check failures so we can address them in a\n  follow-up, making this easier to validate\n\nChange-Id: Ifecb48c7ed761d5170703b556fd6e825be4d2834\nSigned-off-by: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/cd96c9fa7dc95f138b71874bedb37e7474ab5081"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/cd96c9fa7dc95f138b71874bedb37e7474ab5081"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
