)]}'
{"id":"openstack%2Fkeystonemiddleware~654305","triplet_id":"openstack%2Fkeystonemiddleware~master~Ia25abe742628b705e37d895322571e8d2acaeb25","project":"openstack/keystonemiddleware","branch":"master","topic":"bug/1807697","hashtags":[],"change_id":"Ia25abe742628b705e37d895322571e8d2acaeb25","subject":"[PoC] Expose \u0027project_tags\u0027 in auth header","status":"ABANDONED","created":"2019-04-22 05:56:19.000000000","updated":"2019-05-14 01:15:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"32ab280e8f8da8ac11f986dc226f9ccbba21255a","_number":654305,"virtual_id_number":654305,"owner":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-04-22 10:17:11.000000000","_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},{"value":-1,"date":"2019-04-22 10:39:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 10:39:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3704f2530db116fd3d492948af4323a2ccf17d53","author":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"date":"2019-04-22 05:56:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2fbde57a8532d0accbfddd9436b76866d2752bae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 06:34:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/05/654305/1/check/openstack-tox-cover/d9412dd/ : FAILURE in 5m 42s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/654305/1/check/openstack-tox-lower-constraints/77a3627/ : FAILURE in 4m 58s\n- openstack-tox-pep8 http://logs.openstack.org/05/654305/1/check/openstack-tox-pep8/d364379/ : SUCCESS in 5m 37s\n- openstack-tox-py27 http://logs.openstack.org/05/654305/1/check/openstack-tox-py27/e6d69f6/ : FAILURE in 5m 40s\n- openstack-tox-py36 http://logs.openstack.org/05/654305/1/check/openstack-tox-py36/f833787/ : FAILURE in 4m 37s\n- openstack-tox-py37 http://logs.openstack.org/05/654305/1/check/openstack-tox-py37/ef93323/ : FAILURE in 4m 46s\n- openstack-tox-docs http://logs.openstack.org/05/654305/1/check/openstack-tox-docs/be6b7ef/html/ : SUCCESS in 5m 40s\n- tempest-full http://logs.openstack.org/05/654305/1/check/tempest-full/58e50ac/ : FAILURE in 22m 09s\n- tempest-full-py3 http://logs.openstack.org/05/654305/1/check/tempest-full-py3/5209832/ : FAILURE in 22m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d0ff2a7856b7d75671536ec41daaecb8c73a89a0","author":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"date":"2019-04-22 10:05:03.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"61d03f80bae9f5d922b1611eef5097bf843cd3eb","author":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"date":"2019-04-22 10:17:11.000000000","message":"Patch Set 2:\n\nAll test case failed due to the dependency (https://review.opendev.org/#/c/654309)","accounts_in_message":[],"_revision_number":2},{"id":"1c515606a248949d52363c0c34b55b3788b42794","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 10:39:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover http://logs.openstack.org/05/654305/2/check/openstack-tox-cover/5977440/ : FAILURE in 4m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/05/654305/2/check/openstack-tox-lower-constraints/82c212c/ : FAILURE in 4m 00s\n- openstack-tox-pep8 http://logs.openstack.org/05/654305/2/check/openstack-tox-pep8/9ec1938/ : SUCCESS in 4m 40s\n- openstack-tox-py27 http://logs.openstack.org/05/654305/2/check/openstack-tox-py27/9f21754/ : FAILURE in 5m 07s\n- openstack-tox-py36 http://logs.openstack.org/05/654305/2/check/openstack-tox-py36/fc1c1c2/ : FAILURE in 4m 12s\n- openstack-tox-py37 http://logs.openstack.org/05/654305/2/check/openstack-tox-py37/b90192b/ : FAILURE in 5m 14s\n- openstack-tox-docs http://logs.openstack.org/05/654305/2/check/openstack-tox-docs/f221ed3/html/ : SUCCESS in 4m 18s\n- tempest-full http://logs.openstack.org/05/654305/2/check/tempest-full/54b422e/ : FAILURE in 23m 48s\n- tempest-full-py3 http://logs.openstack.org/05/654305/2/check/tempest-full-py3/ade5540/ : FAILURE in 20m 13s","accounts_in_message":[],"_revision_number":2},{"id":"359af95300489546fc75e6d15b04b06a5fca55bc","author":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"date":"2019-05-14 01:15:00.000000000","message":"Abandoned\n\nAuth token should not accept mutable data.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c7908bef81992c644b7ca98555e639c92d4c6b64","revisions":{"9ec143fd6992db9fca98421aa0f03f2092625f43":{"kind":"REWORK","_number":1,"created":"2019-04-22 05:56:19.000000000","uploader":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"ref":"refs/changes/05/654305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/05/654305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/05/654305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/05/654305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/05/654305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/05/654305/1"}}},"commit":{"parents":[{"commit":"4bc09580070c5f6afa9ef39a3d9d1641de557589","subject":"Merge \"Make sure audit middleware use own context\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/4bc09580070c5f6afa9ef39a3d9d1641de557589"}]}],"author":{"name":"Yang Youseok","email":"ileixe@gmail.com","date":"2019-04-22 05:54:00.000000000","tz":540},"committer":{"name":"Yang Youseok","email":"ileixe@gmail.com","date":"2019-04-22 05:54:00.000000000","tz":540},"subject":"[PoC] Expose \u0027project_tags\u0027 in auth header","message":"[PoC] Expose \u0027project_tags\u0027 in auth header\n\nAuth token from Keystone now exposes \u0027project_tags\u0027. Add new header\ntemplate value called \u0027project_tags\u0027 to parse the attribute correctly.\n\nDepends-On: https://review.opendev.org/#/c/654301\nChange-Id: Ia25abe742628b705e37d895322571e8d2acaeb25\nCloses-bug: #1807697\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/9ec143fd6992db9fca98421aa0f03f2092625f43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/9ec143fd6992db9fca98421aa0f03f2092625f43"}]},"branch":"refs/heads/master"},"c7908bef81992c644b7ca98555e639c92d4c6b64":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-04-22 10:05:03.000000000","uploader":{"_account_id":26970,"name":"Yang Youseok","email":"ileixe@gmail.com","username":"ileixe"},"ref":"refs/changes/05/654305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/05/654305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/05/654305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/05/654305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/05/654305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/05/654305/2"}}},"commit":{"parents":[{"commit":"4bc09580070c5f6afa9ef39a3d9d1641de557589","subject":"Merge \"Make sure audit middleware use own context\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/4bc09580070c5f6afa9ef39a3d9d1641de557589"}]}],"author":{"name":"Yang Youseok","email":"ileixe@gmail.com","date":"2019-04-22 05:54:00.000000000","tz":540},"committer":{"name":"Yang Youseok","email":"ileixe@gmail.com","date":"2019-04-22 10:04:44.000000000","tz":540},"subject":"[PoC] Expose \u0027project_tags\u0027 in auth header","message":"[PoC] Expose \u0027project_tags\u0027 in auth header\n\nAuth token from Keystone now exposes \u0027project_tags\u0027. Add new header\ntemplate value called \u0027project_tags\u0027 to parse the attribute correctly.\n\nDepends-On: https://review.opendev.org/#/c/654301\nDepends-On: https://review.opendev.org/#/c/654309\nChange-Id: Ia25abe742628b705e37d895322571e8d2acaeb25\nCloses-bug: #1807697\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c7908bef81992c644b7ca98555e639c92d4c6b64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c7908bef81992c644b7ca98555e639c92d4c6b64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
