)]}'
{"id":"openstack%2Fkeystonemiddleware~677840","triplet_id":"openstack%2Fkeystonemiddleware~master~Ib2c74f4f37855cec250d09b23c45b5b7fde44c8d","project":"openstack/keystonemiddleware","branch":"master","topic":"doc-fix","hashtags":[],"change_id":"Ib2c74f4f37855cec250d09b23c45b5b7fde44c8d","subject":"Comment html_static_path entry in docs conf.py","status":"MERGED","created":"2019-08-21 21:24:06.000000000","updated":"2020-01-08 22:40:24.000000000","submitted":"2019-08-23 19:44:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"677840-1566589470378-e4e6df3c","meta_rev_id":"9f3255af2f378690217cc8c4ec3da12ddcefc575","_number":677840,"virtual_id_number":677840,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":2,"date":"2019-08-23 19:44:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-23 19:04:34.000000000","_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-08-23 19:23:02.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-08-23 19:25:14.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2019-08-23 19:25:14.000000000","_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-23 19:23:02.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-08-23 19:25:14.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-08-23 19:44:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6db0fc2531ad10344474afb44524c64e9478ba6a","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-08-21 21:24:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ec7896caac15bd54c05a87eb85480b314021410","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-21 21:34:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d21cd781f2f4884b12bdbdf890eadab : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0625e6c1528143bea4bd297a8b67741e : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"2e351ae491c54245c01d792b4097ebdcd9756bf0","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-23 16:42:25.000000000","message":"Patch Set 1: Code-Review-1\n\nThe docs job looks green, can you fix the commit message?","accounts_in_message":[],"_revision_number":1},{"id":"dac9732e2aad1a17d72e2977822eb5db5896ed6d","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-08-23 19:04:24.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8a3bfa4544126efd14f2db20255b350c845ea4ea","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-08-23 19:04:34.000000000","message":"Patch Set 2:\n\n\u003e The docs job looks green, can you fix the commit message?\n\nDone!","accounts_in_message":[],"_revision_number":2},{"id":"74e53a2b89e12c85e0c54390afacb3dc18923de4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 19:18:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6db24fe2f01940e899b1a6d9f2da21cd : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d73d30e06004e4d93dfb86e9c9e200c : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":2},{"id":"2bfc09c4ff34b53fba7d4946a82a76c467700680","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-08-23 19:23:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"22a6738eb3b4a146ca3f6e9244c71f955d0917fb","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-08-23 19:25:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"630c209b337ff0854bfe1322c60a250ae72f28d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 19:26:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"17af5f30aa6461df87adeca095371d5d81053d66","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 19:44:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"275bc4e54a7321becb7ebb365e7c14b6005c1a88","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 19:44:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d944be0b17a4a9ab6ed73cc31504be6 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfaf3172017c4c50b57df43864b7a9de : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":2},{"id":"2baf1c8adef478c26fbfef27656e76bba45a6837","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 19:45:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52facb22f9b9483194994142c96714c7 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2},{"id":"ea2058757186bedd0d96719ed53c643543782e89","author":{"_account_id":1916,"name":"Guang Yee","email":"gyee@suse.com","username":"guang-yee"},"date":"2020-01-08 22:40:24.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit ee21aaade9ed595112c1f7d4b64b22bc2015f5ba","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0a65b1420799e7c7f8736e9f6c234f755ab5ac6b","revisions":{"2612c6edfec64c2aabca3fd4e5b62fd09b49b233":{"kind":"REWORK","_number":1,"created":"2019-08-21 21:24:06.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/40/677840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/40/677840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/40/677840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/40/677840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/40/677840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/40/677840/1"}}},"commit":{"parents":[{"commit":"90c6872615273f54c1555aaab5882ba4c12e0309","subject":"Merge \"Bump the openstackdocstheme extension to 1.20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/90c6872615273f54c1555aaab5882ba4c12e0309"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2019-08-21 21:23:41.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2019-08-21 21:23:41.000000000","tz":-300},"subject":"Comment out html static path","message":"Comment out html static path\n\nSee if this fixes the docs gate issue\n\nChange-Id: Ib2c74f4f37855cec250d09b23c45b5b7fde44c8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2612c6edfec64c2aabca3fd4e5b62fd09b49b233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2612c6edfec64c2aabca3fd4e5b62fd09b49b233"}]},"branch":"refs/heads/master"},"0a65b1420799e7c7f8736e9f6c234f755ab5ac6b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-08-23 19:04:24.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/40/677840/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/40/677840/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/40/677840/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/40/677840/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/40/677840/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/40/677840/2"}}},"commit":{"parents":[{"commit":"90c6872615273f54c1555aaab5882ba4c12e0309","subject":"Merge \"Bump the openstackdocstheme extension to 1.20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/90c6872615273f54c1555aaab5882ba4c12e0309"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2019-08-21 21:23:41.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2019-08-23 19:04:22.000000000","tz":0},"subject":"Comment html_static_path entry in docs conf.py","message":"Comment html_static_path entry in docs conf.py\n\nCurrently with sphinx 2.2.0 the docs job is throwing a warning\nthat the html_static_path entry does not exist. We treat warnings\nas errors so this causes the job to fail.\n\nThis change comments the html_static_path entry in conf.py since\nthe path currently does not exist so it appears to be unused.\n\nChange-Id: Ib2c74f4f37855cec250d09b23c45b5b7fde44c8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/0a65b1420799e7c7f8736e9f6c234f755ab5ac6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/0a65b1420799e7c7f8736e9f6c234f755ab5ac6b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
