)]}'
{"id":"openstack%2Fkeystonemiddleware~680624","triplet_id":"openstack%2Fkeystonemiddleware~master~I2b727f28733b1f7e2bb0e2dfa94a39bfdd7449fc","project":"openstack/keystonemiddleware","branch":"master","hashtags":[],"change_id":"I2b727f28733b1f7e2bb0e2dfa94a39bfdd7449fc","subject":"Fix misspell word","status":"MERGED","created":"2019-09-06 08:45:12.000000000","updated":"2019-09-09 04:42:13.000000000","submitted":"2019-09-09 04:41:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"680624-1568004071648-98feb54a","meta_rev_id":"596707c52f4afc1a2056d9aa4f9f4e26d53d3afd","_number":680624,"virtual_id_number":680624,"owner":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":2,"date":"2019-09-09 04:41:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-09 02:46:51.000000000","_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"all":[{"value":0,"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-09-09 02:46:51.000000000","_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-06 09:41:18.000000000","updated_by":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"reviewer":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"state":"REVIEWER"},{"updated":"2019-09-09 02:46:51.000000000","updated_by":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"reviewer":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"state":"REVIEWER"},{"updated":"2019-09-09 04:41:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad1ea2b594854eaf654f367178363e0468b87234","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2019-09-06 08:45:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71bbbeab15d507e447298634125a0110fae3f4c6","author":{"_account_id":27621,"name":"Vishakha Agarwal","email":"agarwalvishakha18@gmail.com","username":"Vishakha"},"date":"2019-09-06 09:41:18.000000000","message":"Patch Set 1: Code-Review-1\n\nTranslations files are auto-generated. These type of changes is irrelevant.","accounts_in_message":[],"_revision_number":1},{"id":"5c34cb96721bd7b053c2b93cfad6de5008c17c13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 10:29:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2da09288a94d4817bd7e55b4666ac36d : SUCCESS in 3m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f8129a440ca410f8da78988e2b38b36 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4126fa8b4a39404186b49827639f7aaf : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4838db4da7104889985313f6ff683a42 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bf3d3a64496447e83bc95f244d15a18 : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9627ae894964893a5b8f7fb51b44cac : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5523481ee14a43dd9a050eaa41381947 : SUCCESS in 5m 47s\n- tempest-full https://zuul.opendev.org/t/openstack/build/706e4be5ef994cc58fd6d8dcac000b90 : SUCCESS in 1h 30m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e6ed85e83214257931a8f25db9ee67a : SUCCESS in 1h 36m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed2610c7a693431dbe1d00d70018d6cb : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a5165042e4ac07260034964f1162f507ea8e548c","author":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"date":"2019-09-09 01:02:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b150c3f357369499e6f11244dc2ed3f2ba1f9624","author":{"_account_id":15054,"name":"wangxiyuan","email":"wangxiyuan1007@gmail.com","username":"wangxiyuan"},"date":"2019-09-09 02:46:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fb54c46c116b7319faaa0e8eb3e1e732a7d909bc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 02:51:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7eaeb436bb4a43989a56caf28ce1abb4 : SUCCESS in 4m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4984508cd7b4b0b88ed12bd67090fef : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3642e7007c4141698cf40e6c9cafa554 : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9484122387f04745a742d0c520f3a109 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2538b553f8824a17a65eaa0b8ca23672 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/906987ea89414a6fbb32af1d1da71af4 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54c4a8aaa0344a6e8269bc73fa8a246b : SUCCESS in 4m 20s\n- tempest-full https://zuul.opendev.org/t/openstack/build/cb69de7ac33449dfa6abd349e47f62f1 : SUCCESS in 1h 46m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e18700e577f4976968212b3f92886a5 : SUCCESS in 1h 41m 19s","accounts_in_message":[],"_revision_number":2},{"id":"5d5f58f9264218c60f784e060e6c5f714fa22043","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 02:52:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4ac1b718eefa540c83ec9fccdb4eb1d0fc74388f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 04:41:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"55b5b837c25b87d4a8e9f66aff0b06d8df207e11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 04:41:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9129dd9532fa44ebb13cc80c0d8d1a85 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abdbd0624950453ea12e05eed8a1d83f : SUCCESS in 2m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2dfa324c00b248bc850f648869a554ba : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cac805a9bd445749ecb3c401029b330 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ecc5aba95a9417fa8f4e8cd4b301f16 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5c83db7e1304d44b8661f62e33621c0 : SUCCESS in 4m 26s\n- tempest-full https://zuul.opendev.org/t/openstack/build/7b826683efbd4cea9c6c8479e083500c : SUCCESS in 1h 47m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08726acc89a24ed9a4372b67de30bf7e : SUCCESS in 1h 23m 05s","accounts_in_message":[],"_revision_number":2},{"id":"596707c52f4afc1a2056d9aa4f9f4e26d53d3afd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 04:42:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11af1114ca72473cb6ed6ce5676fbcfd : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d1bd783a77e5b6ae057ba9b13833ff0584e1589b","revisions":{"b241db545c7e32a335ffc7e771d06bcc944f82d8":{"kind":"REWORK","_number":1,"created":"2019-09-06 08:45:12.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/24/680624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/24/680624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/24/680624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/24/680624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/24/680624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/24/680624/1"}}},"commit":{"parents":[{"commit":"0a65b1420799e7c7f8736e9f6c234f755ab5ac6b","subject":"Comment html_static_path entry in docs conf.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/0a65b1420799e7c7f8736e9f6c234f755ab5ac6b"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-06 08:44:19.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-06 08:44:58.000000000","tz":480},"subject":"Fix misspell word","message":"Fix misspell word\n\nChange-Id: I2b727f28733b1f7e2bb0e2dfa94a39bfdd7449fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/b241db545c7e32a335ffc7e771d06bcc944f82d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/b241db545c7e32a335ffc7e771d06bcc944f82d8"}]},"branch":"refs/heads/master"},"d1bd783a77e5b6ae057ba9b13833ff0584e1589b":{"kind":"REWORK","_number":2,"created":"2019-09-09 01:02:30.000000000","uploader":{"_account_id":30408,"name":"Qitao","email":"qihongtao@inspur.com","username":"qihongtao"},"ref":"refs/changes/24/680624/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/24/680624/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/24/680624/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/24/680624/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/24/680624/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/24/680624/2"}}},"commit":{"parents":[{"commit":"0a65b1420799e7c7f8736e9f6c234f755ab5ac6b","subject":"Comment html_static_path entry in docs conf.py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/0a65b1420799e7c7f8736e9f6c234f755ab5ac6b"}]}],"author":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-06 08:44:19.000000000","tz":480},"committer":{"name":"Q.hongtao","email":"qihongtao@inspur.com","date":"2019-09-09 01:02:14.000000000","tz":480},"subject":"Fix misspell word","message":"Fix misspell word\n\nChange-Id: I2b727f28733b1f7e2bb0e2dfa94a39bfdd7449fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d1bd783a77e5b6ae057ba9b13833ff0584e1589b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d1bd783a77e5b6ae057ba9b13833ff0584e1589b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
