)]}'
{"id":"openstack%2Fkeystonemiddleware~695809","triplet_id":"openstack%2Fkeystonemiddleware~master~Ie00fc310728c319faf2cfdfb651f0c7a8f48d757","project":"openstack/keystonemiddleware","branch":"master","topic":"remove-ksc","hashtags":[],"change_id":"Ie00fc310728c319faf2cfdfb651f0c7a8f48d757","subject":"Remove keystoneclient exception usage in tests","status":"MERGED","created":"2019-11-24 06:14:21.000000000","updated":"2020-01-02 18:58:56.000000000","submitted":"2020-01-02 18:58:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695809-1577991536510-d94f54cc","meta_rev_id":"60f8e61f765ffaeb7863df85fdaf44597608399c","_number":695809,"virtual_id_number":695809,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-02 18:58:56.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"all":[{"value":2,"date":"2019-12-10 18:57:32.000000000","_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2020-01-02 17:09:57.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":1,"date":"2020-01-02 17:09:57.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-10 18:57:32.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2020-01-02 17:09:57.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-01-02 18:58:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ed93b480f0f38da4b05c0d4e3c2f880025253c5","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2019-11-24 06:14:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5542e717c0e0d7f019f10e786cf6d1b239f83276","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-24 07:54:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a439b924ca7249d69b3e660475b13e19 : SUCCESS in 4m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c3282c0b85242bdbe6bcd6cb68a9849 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af9f5a3bcb714bbba4bdd1d5616568e8 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b0d1eaf6ea0a4cb9a85394cac7107a70 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9953caf9db344dd7850dd213c60d859b : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7e8a60ea77a1454ead66ca0d9a37402e : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a835ec26b15417ba97fc4e2ffdc3e38 : SUCCESS in 5m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c320f7cb11f845e9b0ab41eb065e8331 : SUCCESS in 1h 29m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4453d56d23044b948108bbd5e7f2de64 : SUCCESS in 1h 32m 48s","accounts_in_message":[],"_revision_number":1},{"id":"a94e0ff30ec0d34a4cf39fdfb2645ee846ea88a3","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-12-10 18:57:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e6cb12da2c2a576fd6ec52d83499ffcfcc74fd2d","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-01-02 17:09:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fc435a7415117fc7f5ee9d922e6820aecf8fe2f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 17:10:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"68f77cc61817b1c1e17aa875db31c98f9f1dd906","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 18:58:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"80359d12bce06f95f0e37d224d761a90f389ecc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-02 18:58:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca53189320194a288d07ad5a2ebe0280 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eee04d586c9e434cbb992e2876f4a649 : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aff7b7dd75944020913dc5bca98536ce : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b07c567d3bd747cea580c5e46c2de76d : SUCCESS in 3m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/11527e3417514cf39b071e3f7b138c30 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99e12921390f4c01a58939eec7f9e5c1 : SUCCESS in 5m 15s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6fcf03ce54e74a20ae9fc5d98d7e428d : SUCCESS in 1h 43m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bafea7030e6a472eb211905edd4721bb : SUCCESS in 1h 23m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3090bfbc0adffff36b158785a1ecea73349045c","revisions":{"d3090bfbc0adffff36b158785a1ecea73349045c":{"kind":"REWORK","_number":1,"created":"2019-11-24 06:14:21.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/09/695809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/09/695809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/09/695809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/09/695809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/09/695809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/09/695809/1"}}},"commit":{"parents":[{"commit":"19f2791082f906132c65770e50334648d8f0c6c0","subject":"Merge \"Generate pdf documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/19f2791082f906132c65770e50334648d8f0c6c0"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2019-11-24 06:12:24.000000000","tz":-360},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2019-11-24 06:12:24.000000000","tz":-360},"subject":"Remove keystoneclient exception usage in tests","message":"Remove keystoneclient exception usage in tests\n\nThis change replaces the usage of keystoneclient exceptions in the\nauth_token_middleware unit tests to use the ConnectFailure exception\nfrom keystoneauth.\n\nThis is part of the process of removing keystoneclient from\nkeystonemiddleware.\n\nChange-Id: Ie00fc310728c319faf2cfdfb651f0c7a8f48d757\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d3090bfbc0adffff36b158785a1ecea73349045c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d3090bfbc0adffff36b158785a1ecea73349045c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
