)]}'
{"id":"openstack%2Fkeystonemiddleware~702278","triplet_id":"openstack%2Fkeystonemiddleware~master~I8799d9318776621417024ed4251e4e00944e951a","project":"openstack/keystonemiddleware","branch":"master","topic":"ksc-removal","hashtags":[],"change_id":"I8799d9318776621417024ed4251e4e00944e951a","subject":"WIP - Remove ksc from ksm","status":"ABANDONED","created":"2020-01-13 18:38:09.000000000","updated":"2020-07-15 20:52:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"eca04dad4b84087939d926099741a8de71918948","_number":702278,"virtual_id_number":702278,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-18 15:42:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-18 15:42:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"361e7648f878df98c787da6d7f6642c68ee3d44c","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-01-13 18:38:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b91d5afa5d85166d48b1b4c5535522432c46896","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 19:05:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b03110190edf4e2788f5c0aaa9cbfcf9 : FAILURE in 3m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f19f6cb1cbde4dbc9d329d671584f6f1 : FAILURE in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/459864fee72240ae9128cce2bb6b6bc7 : FAILURE in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2895ab69c8e4e76af1444e6e2e09482 : FAILURE in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6d5815f67a34d85b458adff0a65938a : FAILURE in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5e19270eae84487b08e27f1b14fd3d6 : SUCCESS in 4m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42079daa3b564ded9201199b30505bdd : FAILURE in 25m 47s","accounts_in_message":[],"_revision_number":1},{"id":"33742146b309a9418d28e8df3623197c96b423e7","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-03-18 15:07:26.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"92ae2e2808bbf5b907bc267a34f5535c03121356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 15:42:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61e6234cb1de4774be5523a8e0787efd : FAILURE in 4m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dcae22ae6d974407b31689395de4f039 : FAILURE in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5d9f1179b3c4bcb902a1ed838fdc914 : FAILURE in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/462510547687460f829da736afb8d6eb : FAILURE in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/698142e588aa44b88f67665df28d2ce1 : FAILURE in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b1dff9f2dd24a3189ef4ed3016a77ae : FAILURE in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2a70d4c5c6c41c2b4b6dc9bd8da5bcd : SUCCESS in 5m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e1371c226894f16819c778dedd23b17 : FAILURE in 26m 17s","accounts_in_message":[],"_revision_number":2},{"id":"858c522bcdc9ca962eae4f4623f665ade9e0cc63","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-07-15 20:52:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5a9e33a61777827ec181c8a626c827540a8f2026","revisions":{"054de1c1817cc1ef7c2bde088337daae7e5114da":{"kind":"REWORK","_number":1,"created":"2020-01-13 18:38:09.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/78/702278/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/78/702278/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/702278/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/702278/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/702278/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/702278/1"}}},"commit":{"parents":[{"commit":"bf397330179234a0df9e08c7d201175b046faf6a","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/bf397330179234a0df9e08c7d201175b046faf6a"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2020-01-13 18:37:50.000000000","tz":-360},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2020-01-13 18:37:50.000000000","tz":-360},"subject":"WIP - Remove ksc from ksm","message":"WIP - Remove ksc from ksm\n\nChange-Id: I8799d9318776621417024ed4251e4e00944e951a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/054de1c1817cc1ef7c2bde088337daae7e5114da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/054de1c1817cc1ef7c2bde088337daae7e5114da"}]},"branch":"refs/heads/master"},"5a9e33a61777827ec181c8a626c827540a8f2026":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-18 15:07:26.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/78/702278/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/78/702278/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/702278/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/702278/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/702278/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/78/702278/2"}}},"commit":{"parents":[{"commit":"66352680ab7abc1c2bcc0dbe61c5558c41e4ff20","subject":"Remove universal wheel configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/66352680ab7abc1c2bcc0dbe61c5558c41e4ff20"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2020-01-13 18:37:50.000000000","tz":-360},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2020-03-18 15:07:26.000000000","tz":0},"subject":"WIP - Remove ksc from ksm","message":"WIP - Remove ksc from ksm\n\nChange-Id: I8799d9318776621417024ed4251e4e00944e951a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5a9e33a61777827ec181c8a626c827540a8f2026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5a9e33a61777827ec181c8a626c827540a8f2026"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
