)]}'
{"id":"openstack%2Fkeystonemiddleware~751055","triplet_id":"openstack%2Fkeystonemiddleware~stable%2Fvictoria~I8aceffbc57970a2b8498a396caee4bb5a26d8b34","project":"openstack/keystonemiddleware","branch":"stable/victoria","topic":"create-victoria","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2023-06-05 13:14:58.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-01-26 16:31:28.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2023-10-06 14:24:59.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}},"hashtags":[],"change_id":"I8aceffbc57970a2b8498a396caee4bb5a26d8b34","subject":"Update .gitreview for stable/victoria","status":"MERGED","created":"2020-09-10 17:06:14.000000000","updated":"2024-01-26 16:32:22.000000000","submitted":"2024-01-26 16:31:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751055-create-victoria","meta_rev_id":"11c648720d0076418be8820ee257414e8b8f6284","_number":751055,"virtual_id_number":751055,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-26 16:31:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-08-04 14:04:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2023-10-06 14:24:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2023-08-04 14:04:52.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2024-01-26 14:53:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-10 17:07:10.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-10 17:35:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-07 15:30:34.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"CC"},{"updated":"2022-05-13 04:37:14.000000000","updated_by":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"reviewer":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"state":"REMOVED"},{"updated":"2022-07-15 15:33:16.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2023-06-02 14:17:31.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"354919e6366593a16e9b0dacf9061cfc084c72b1","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-10 17:06:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bbfb724ab2893cb686dd14568d5c99a979eebd2a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-10 17:07:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"589d7490c7fb12a1323657fc59146d645d81c83e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 17:35:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b5b74d0e0304151b4549aeb906087c9 : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d821968fc716428ebd0d356d6f1c29c3 : FAILURE in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e97cdaf20bcf40c8870dadc12ceeb278 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/caf3381fb52d4d45811ee53c7f13f355 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c37157150c1a4b53802173211937c222 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bb62cc364404e6bb0420c122f5f85bd : SUCCESS in 18m 59s","accounts_in_message":[],"_revision_number":1},{"id":"c48bc498aacfcfd1da04af60c83d7de78cbdad5a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-21 18:16:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8bd34f61184c50f952f0b37db637414776cab9a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-21 18:43:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e26238d8f054028b4184a4d91fe7298 : SUCCESS in 3m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/130fd7190dc249c69931ce02066f1a7b : FAILURE in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4285ddec67d740aabed8ff0e265b52cc : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2aa82613e1ec4afabd4f6846b98379ba : SUCCESS in 5m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/068657d377354a0ebbe37e1edb80d765 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97b5a395f04542979d6aa5665f259e7e : SUCCESS in 13m 31s","accounts_in_message":[],"_revision_number":1},{"id":"40faf8b71c3e66612083b883cdcaa914c78b0919","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-12-04 19:25:36.000000000","message":"Patch Set 1: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7f1ed0a9db6d8bf4e9290ffc0a4061477c6837a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-04 20:51:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e16c055c52fd4c8c94ec4de91bea33ec : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c277d46348c64370a57a0ffade278a60 : FAILURE in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6afa5758a5d41088936d8d2e1606991 : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3e20591334b4b888a16809576a50fc5 : SUCCESS in 3m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da4d54ad0bb94fe39088b9fe3741a209 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d79c240e7c94536a479624b0cf1c00e : SUCCESS in 5m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80008cdd53074d3ea9d3a1da2e75608f : SUCCESS in 1h 24m 09s","accounts_in_message":[],"_revision_number":1},{"id":"ee687e5e32106f1dfe8166437e95aa11874b989b","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-02-07 15:30:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eccd316133f35242c15265a458989f0c75e42abc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 17:07:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b1b8cc9b53e429b8d3541c875a8bf39 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32b021e701d248ad8134a082a105ca4c : FAILURE in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/436c8c8cd39f436d8e3dd33b5e4a6ce1 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2adff32dc7b4160b053f7b34343546d : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bc2b5ef06f3048e6b4fb0850bcd455bc : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76a71f045fb4441080cb04c931241a73 : SUCCESS in 6m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/deb1b49ffdb948e19fe31a309cc7a573 : SUCCESS in 1h 24m 07s","accounts_in_message":[],"_revision_number":1},{"id":"d208294d01f35b6a379c22b826da908c18176bfe","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2022-05-13 04:37:14.000000000","message":"Removed cc Gage Hugo.","accounts_in_message":[],"_revision_number":1},{"id":"b51a0c1bb9e8badff5aa65719f25224ebf286f2e","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2022-07-15 15:33:16.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"261df4b26c81649090946ca5e3ae2d4bcab1eee2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 16:50:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/72903a1f568d4dbd8d3485d9d019466f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21465dc266f641fea3d16f2a82da35d8 : SUCCESS in 6m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35a25e4d4ba14d89a3fa062b8231c814 : FAILURE in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/333d9380f3714a35b806213db02dbae6 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b63fd773cb04710b7b130394e3527b2 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5aa92d7357e942c7a1e0302ace45913a : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f4fa8ec5c424857a68e53b6915f6a29 : SUCCESS in 6m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36b119ecc0014e5583bec11f2b556ed7 : FAILURE in 1h 15m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a8866789c82d17a13cbd1c573d33b3444b9e7568","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2023-06-02 14:16:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"35a03b1164ef164ae8fee4676274d9445342e27b","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-06-02 14:17:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f165f2304e1d6316647dfadf783fedfafefbc884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-02 15:35:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3f9613fa7cb464db35bf50dab5aa2cd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b1297e5e937475698fe609e4df5c4f6 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb837e8de1f14e8a8bc2b906e8d15b64 : SUCCESS in 2m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a84f015d5c74e5897b27ab535d55223 : SUCCESS in 3m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/704c31acaf4a41f9a9c98a97b8603ec7 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65da527f2d6d42eca62dd59a30b63b47 : SUCCESS in 4m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b25b16a0cf5e48be859cfda8289c9603 : FAILURE in 1h 16m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85ab4a46c13c420889ac2edb93358b06 : FAILURE in 3m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"225484b44a48fcd1f672afc8f18978711ca6c6e8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-06-05 13:14:58.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a7d21d557fb43bea4cd1986ecbc330ccb23c269d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 14:42:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b14e006f6304f95949fd8d2c4c505bf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3010174af012476fb21b68a2bfd7071e : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beab755dfe1c4b1f9bc591fb78aa7434 : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0782172c843b44eb8a9890ae3819a05c : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd46158f48f34fd980db0f4d65d0a06d : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e3ddc04480436fb4f3f7d2acf8e19d : SUCCESS in 6m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/579548fe77bf4a09b1ecb1cff45d6c34 : SUCCESS in 1h 26m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4bd127f5fe64cb1b2bdfa7dbc13b995 : FAILURE in 1m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5de0e7358859a3004b08e1ccba4e39de77ec71cc","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-07-17 13:12:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f9892b916c07af971d61b4aeefd752c87615c68d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 13:13:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"733005bf0dd8438a3d647409e2ad651e26b830c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 14:37:31.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2476dc6f6994d6dbc464f6716b42f5e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1585d7736d342e4bc889c5002f22d38 : FAILURE in 2m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/217eb2d2e141485a9a0c9763ec3f7ed6 : SUCCESS in 3m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9345b64db9b648dda17c90500f33c8ac : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f7c0ed10ce14dc181b2d852fef29490 : SUCCESS in 6m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02ae859c1f6c46d984d4f5897903eb86 : SUCCESS in 1h 21m 02s","accounts_in_message":[],"_revision_number":2},{"id":"749f1b775a206cb2c0a36391b24345549dae4f46","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2023-07-21 14:48:23.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"46f95213d35516e6b2a1e50b560049e981cf5286","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2023-08-04 14:04:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3f7ba6ecbbd000d0c12ce64bb96a5103c599e5fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 15:23:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14494e389a4c49ba8728f20e7814a975\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/670e3c47d75948338cfdea43249cfd31 : FAILURE in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/325bf86170ac40e4ba2f50417765ec12 : FAILURE in 2m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f74fc2d74934119a8177cc101d1a362 : SUCCESS in 2m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9e4af4f5ec284fa59350872a2a6a2770 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efcda53c8f0b4372b11c6f19f803cbe7 : SUCCESS in 5m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e0ca2c28c52a4fab9efb46072bebc2b9 : SUCCESS in 1h 17m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32a6d01a344d4b5e92c1a3de5aeac681 : FAILURE in 2m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0ba9b3fbc545b3539ebeb4ddde00028595139a1b","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2023-10-06 14:24:59.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cd2ad6fa3b205026b448fb964c4bca8af26f3b11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 16:06:51.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce191fbb22954047aaee9e581b837bad\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/661127b30daf4c549e6c86933ce83156 : FAILURE in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d36f6f02881b4ce99b2c6dae32ab5703 : FAILURE in 2m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c61fe4376e39433eae1bd39d2114dc5c : SUCCESS in 3m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09b3ccd863244bb9ab3adfee86a9d9e0 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3278570da36943d7a2201c4db20813c0 : SUCCESS in 5m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cbaae8aa2000430e99a296e09c11221c : SUCCESS in 1h 34m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6652857a27446f88131173df3a7a3ef : FAILURE in 2m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"438d0ce6ac41929a5136c6c3a03278ead49b66c2","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-01-05 15:37:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a6b53423c7aafb8564b46c984c7ddd785a1b67da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 16:57:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3d8b534a6464b5d843fd1c301737d9e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f257c719ace14d3abb9deac9a1dc0e63 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdf2a6e27a1248de92d83bea95179e78 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22988f8ff9b44432929680c9d56d2c15 : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/020bb812cb7a42b9b286332575a29730 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c75107251cc4eeab6af0a70b0465777 : SUCCESS in 6m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4852d366525547308156bdfdeeb2ed69 : SUCCESS in 1h 11m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/587b6589b2a14fb29fa02d6985f7c1b1 : FAILURE in 3m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6bc5aca3b72e30eb001a60d6df6d369a8a5313eb","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-01-26 14:53:16.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0500129b95e085fcde7a3a625deb1db1918c84b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 14:53:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3259a228d97b55d17baca6dd8c7f1279321b8eb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 16:31:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4be84f50f0fd4697ae59bb07f5beccbd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e5431ac38e04052a200e41d4a26dad4 : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b21b2353d1541e58ab0277f0ed84250 : SUCCESS in 3m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2de9f4db80314f0fa7b6785bf320df9d : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5f314296a384109ad9d0b583100421a : SUCCESS in 5m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/072dd076072642a283c8ce0ede692d43 : SUCCESS in 1h 24m 59s","accounts_in_message":[],"_revision_number":3},{"id":"5e21bde660aa1c84df41d0e63953102943a22615","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 16:31:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"11c648720d0076418be8820ee257414e8b8f6284","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 16:32:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca61f4f5eab14ed6b43e17671160f133\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7295911ac1874f36902e9e967b3fe418 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"27dda2ba8ea1facae4ceb4ac4e76428af74181f8","revisions":{"423541dc310780ea30405142850e387ae8d71cc6":{"kind":"REWORK","_number":1,"created":"2020-09-10 17:06:14.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/55/751055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/55/751055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/55/751055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/55/751055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/55/751055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/55/751055/1"}}},"commit":{"parents":[{"commit":"654d31a1bd32968bfa008a117a60b58e18d68df8","subject":"Merge \"Change the default Identity endpoint to internal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/654d31a1bd32968bfa008a117a60b58e18d68df8"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-10 17:06:10.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-10 17:06:10.000000000","tz":0},"subject":"Update .gitreview for stable/victoria","message":"Update .gitreview for stable/victoria\n\nChange-Id: I8aceffbc57970a2b8498a396caee4bb5a26d8b34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/423541dc310780ea30405142850e387ae8d71cc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/423541dc310780ea30405142850e387ae8d71cc6"}]},"branch":"refs/heads/stable/victoria"},"f84accf215bfd56ea5fcdb73d311b3c842e789de":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-02 14:16:44.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/55/751055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/55/751055/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/55/751055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/55/751055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/55/751055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/55/751055/2"}}},"commit":{"parents":[{"commit":"44b69c71a014c97ebbe128f56deea882c26ef4e1","subject":"Fix several errors in CI","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/44b69c71a014c97ebbe128f56deea882c26ef4e1"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-10 17:06:10.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2023-06-02 14:16:44.000000000","tz":0},"subject":"Update .gitreview for stable/victoria","message":"Update .gitreview for stable/victoria\n\nChange-Id: I8aceffbc57970a2b8498a396caee4bb5a26d8b34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f84accf215bfd56ea5fcdb73d311b3c842e789de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/f84accf215bfd56ea5fcdb73d311b3c842e789de"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"},"27dda2ba8ea1facae4ceb4ac4e76428af74181f8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-08-04 14:04:52.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/55/751055/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/55/751055/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/55/751055/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/55/751055/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/55/751055/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/55/751055/3"}}},"commit":{"parents":[{"commit":"4f2bb39a5d79f5f8c2e45f3ae782fcb48c55cd27","subject":"Remove cache invalidation when using expired token","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/4f2bb39a5d79f5f8c2e45f3ae782fcb48c55cd27"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-10 17:06:10.000000000","tz":0},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2023-08-04 14:04:52.000000000","tz":0},"subject":"Update .gitreview for stable/victoria","message":"Update .gitreview for stable/victoria\n\nChange-Id: I8aceffbc57970a2b8498a396caee4bb5a26d8b34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/27dda2ba8ea1facae4ceb4ac4e76428af74181f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/27dda2ba8ea1facae4ceb4ac4e76428af74181f8"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
