)]}'
{"id":"openstack%2Fkeystonemiddleware~768432","triplet_id":"openstack%2Fkeystonemiddleware~master~Id62149ddfc52a9ab7f6224813434152bc282d3cf","project":"openstack/keystonemiddleware","branch":"master","hashtags":[],"change_id":"Id62149ddfc52a9ab7f6224813434152bc282d3cf","subject":"Use TOX_CONSTRAINTS_FILE","status":"ABANDONED","created":"2020-12-24 07:06:18.000000000","updated":"2021-03-05 05:56:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"669e6367ff9ec4cb648a36d01052d7e844412e66","_number":768432,"virtual_id_number":768432,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-24 08:28:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-24 08:28:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60dcaa83c1ed18440558c7451d97fc03246cf2f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2020-12-24 07:06:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23ad4703ce5951d0830a967b8dc2e1cd506a3a21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 08:28:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14b9ecdf6a4c4a168a863b669c808567 : SUCCESS in 4m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42c8e9d06405439092c89d0e2fd9f93f : FAILURE in 2m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8641f42002ae4345a9f1eb761acb073d : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63524eb04eb74b479015878243796a27 : SUCCESS in 3m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/926e2faf787a44f78e59b384e5938956 : SUCCESS in 3m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8cee8b50a7484e59be5dd874c2920ee9 : SUCCESS in 2m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8679a592ce484c25b62b2fe0ba3f8782 : SUCCESS in 5m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3cded465677422b9b8f987356269797 : SUCCESS in 1h 20m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2344f09d61664242baecc512094e7553 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"669e6367ff9ec4cb648a36d01052d7e844412e66","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2021-03-05 05:56:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c0daf51f9bc72566e5ccd6876a08b85e04e98044","revisions":{"c0daf51f9bc72566e5ccd6876a08b85e04e98044":{"kind":"REWORK","_number":1,"created":"2020-12-24 07:06:18.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/32/768432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/32/768432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/32/768432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/32/768432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/32/768432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/32/768432/1"}}},"commit":{"parents":[{"commit":"3659bdad861c1648dc53a1f66c968e82b498eb9c","subject":"Merge \"[goal] Migrate testing to ubuntu focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/3659bdad861c1648dc53a1f66c968e82b498eb9c"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-12-24 07:05:20.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2020-12-24 07:06:13.000000000","tz":480},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\nThis allows to use upper-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\nChange-Id: Id62149ddfc52a9ab7f6224813434152bc282d3cf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c0daf51f9bc72566e5ccd6876a08b85e04e98044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c0daf51f9bc72566e5ccd6876a08b85e04e98044"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
