)]}'
{"id":"openstack%2Fkeystonemiddleware~777554","triplet_id":"openstack%2Fkeystonemiddleware~stable%2Fussuri~I0e96334b65a0bf369ebf1d88651d13feb8d2ecac","project":"openstack/keystonemiddleware","branch":"stable/ussuri","topic":"bug/1883659","hashtags":[],"change_id":"I0e96334b65a0bf369ebf1d88651d13feb8d2ecac","subject":"Switch to eventlet-safe oslo.cache\u0027s MemcacheClientPool","status":"ABANDONED","created":"2021-02-26 14:05:03.000000000","updated":"2021-04-11 07:40:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":773939,"cherry_pick_of_patch_set":8,"meta_rev_id":"625465bb8fa58e1193481ed0d1efec149fea5033","_number":777554,"virtual_id_number":777554,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-24 21:31:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-26 15:41:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-23 15:58:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"}],"messages":[{"id":"1ff2d13dcbfc60bb7b7529473b40862325f6b3fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-02-26 14:05:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"15141dcab3a3a0678d8c6edb7f2a89fc8c61a847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 15:41:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/419cabc2cf88463484505309fec5f436 : SUCCESS in 3m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2cbef34a289411fa18b9bd3b81a2240 : FAILURE in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59a84db3883f41099fd950c22ecc1564 : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96be205147f74ad39479da81b47b7c65 : SUCCESS in 3m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b42ef3d39a443bab5ea89ad7168339e : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cacada01fdcd4c959ecd53bf16401281 : SUCCESS in 5m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/088da7b4c93941e2a76ce8d587d7208a : SUCCESS in 1h 30m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d619c007a5634619a8c9a86d78b15555 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"5c113dbc90d827cc5bd6b670665a1c257c56f699","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-02-26 15:50:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b9a93d5195d61198bf76d7c09af847a800b38865","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 17:25:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/454a3cb33176455a8b0ed42eca33e50c : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/372f721655224a6ea6492841b6a2aece : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b01fe4a59ecf41fe8eeb57a28e6f6c5a : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5f495fd7bc4449e9403072e2385b51d : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf55b22fd9cc4924896478325f846dec : SUCCESS in 5m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae889ca38ac648b2a1ed66a4fb478630 : SUCCESS in 1h 20m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19de888d62394dc1bacb8ef89e045f14 : SUCCESS in 4m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ac32b97a4e3466d8deecdafd8a108e9 : FAILURE in 3m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"df2040f38e5fa3b3a5bc8affe4c1c399d4df15d5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-23 15:58:08.000000000","message":"Patch Set 2:\n\nPlease can you `cherry-pick -x` these changes (from victoria) to allow to retrieve the backports history.","accounts_in_message":[],"_revision_number":2},{"id":"d532cf5d98c093f869d4199aba36d9f1d75e50b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-24 19:50:38.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7421169970c42509ca51e9553e8d4fdda84a487d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-24 20:01:52.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Please can you `cherry-pick -x` these changes (from victoria) to allow to retrieve the backports history.\n\nDone, but cherry-picked from master, it\u0027s not merged to victoria","accounts_in_message":[],"_revision_number":3},{"id":"1ac5837f6691fe1e1d42eb91b02a90c13f6de602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 21:31:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4d872f8fda646ec9b4ac8b5aa37c521 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e9b68d634a44e269e530143afeb04ad : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/550296a872c04d15acfa77c9ee497fb8 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1762e24b96e54618b7b4918d2bd68b6e : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b97c848bd15d473ea84e74ffc86d66a3 : SUCCESS in 6m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/154732faa05c4a6598994beb655e6cf8 : SUCCESS in 1h 33m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fe159dcd5574b0ba891f452d1cffd97 : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3607a6605dd84907b5552e4096fd2262 : FAILURE in 3m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b617471c98307036ec96265b43b4a4748102eb9c","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-04-11 07:40:45.000000000","message":"Patch Set 3:\n\nno action from upstream, let\u0027s abandon this","accounts_in_message":[],"_revision_number":3},{"id":"625465bb8fa58e1193481ed0d1efec149fea5033","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-04-11 07:40:49.000000000","message":"Abandoned\n\nno action from upstream, let\u0027s abandon this","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3ae39554c32c195a1549f4079e2ea1224356003c","revisions":{"250508beb07095127c67413d700670ef137368e3":{"kind":"REWORK","_number":1,"created":"2021-02-26 14:05:03.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/54/777554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/54/777554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/54/777554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/54/777554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/54/777554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/54/777554/1"}}},"commit":{"parents":[{"commit":"113b41193c2213e572c8da2d1f2ed6849843dc44","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/113b41193c2213e572c8da2d1f2ed6849843dc44"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-02-10 13:47:17.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-02-26 14:05:03.000000000","tz":0},"subject":"Switch to eventlet-safe oslo.cache\u0027s MemcacheClientPool","message":"Switch to eventlet-safe oslo.cache\u0027s MemcacheClientPool\n\nIn past days there were discussions about various issues\nwith memcached connections [1][2][3].\n\nAfter investigation it looks like common root cause for above\nproblems is keystonemiddleware. More precisely said the way\nhow keystonemiddleware is caching tokens.\n\nCurrently it\u0027s using some home-made CachePool with direct\nusage of memcached library, moreover it looks like its\napproach is not eventlet-safe.\nDiscussion can be mainly found in [4].\n\nFortunately keystonemiddleware can use \"advanced cache pool\",\nwhich is oslo.cache\u0027s implementation and was added long time ago [5],\nbut it is turned on only if memcache_use_advanced_pool\u003dTrue.\n\nThis patch is switching to more elaborated oslo.cache CachePool\nand adding deprecation warning about eventlet-unsafe variant\nof keystonemiddleware\u0027s memcache pool.\n\nHow to reproduce ?\n\nwith memcache_use_advanced_pool\u003dFalse\n\n1. Build clean ENV of openstack\n2. Deploy core projects (keystone,glance,nova,placement...)\n3. Run while true; do COMMAND FOR SERVICE; done\n   - several bashes, in parallel (5-7)\n\nCOMMAND FOR SERVICE:\n- openstack network list\n- openstack volume list\n- openstack server list\n- openstack image list\n\n4. Check memcached connections (which will grow up):\n    - ss | grep 11211 | wc -l   every second\n\nHow to fix and test it ?\n\nRepeat above, to fix:\n - with memcache_use_advanced_pool\u003dTrue\n   OR\n - apply this patch\n\nCompare measurements in graph.\n\n[1] https://bugs.launchpad.net/keystonemiddleware/+bug/1892852\n[2] https://bugs.launchpad.net/oslo.cache/+bug/1888394\n[3] https://bugs.launchpad.net/keystonemiddleware/+bug/1883659\n\n[4] https://review.opendev.org/c/openstack/oslo.cache/+/742193\n\n[5] https://review.opendev.org/c/openstack/keystonemiddleware/+/268664\n\nCloses-Bug: #1883659\nCloses-Bug: #1892852\nCloses-Bug: #1888394\n\nChange-Id: I0e96334b65a0bf369ebf1d88651d13feb8d2ecac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/250508beb07095127c67413d700670ef137368e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/250508beb07095127c67413d700670ef137368e3"}]},"branch":"refs/heads/stable/ussuri"},"6b64b0841b385e18f8f889fa9f23a3e7a2edb0c1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-26 15:50:55.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/54/777554/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/54/777554/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/54/777554/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/54/777554/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/54/777554/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/54/777554/2"}}},"commit":{"parents":[{"commit":"5f540f7c1f5af45131c2d48079ef6275081c1fe8","subject":"Updating lower-constraints job as non voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5f540f7c1f5af45131c2d48079ef6275081c1fe8"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-02-10 13:47:17.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-02-26 15:50:55.000000000","tz":0},"subject":"Switch to eventlet-safe oslo.cache\u0027s MemcacheClientPool","message":"Switch to eventlet-safe oslo.cache\u0027s MemcacheClientPool\n\nIn past days there were discussions about various issues\nwith memcached connections [1][2][3].\n\nAfter investigation it looks like common root cause for above\nproblems is keystonemiddleware. More precisely said the way\nhow keystonemiddleware is caching tokens.\n\nCurrently it\u0027s using some home-made CachePool with direct\nusage of memcached library, moreover it looks like its\napproach is not eventlet-safe.\nDiscussion can be mainly found in [4].\n\nFortunately keystonemiddleware can use \"advanced cache pool\",\nwhich is oslo.cache\u0027s implementation and was added long time ago [5],\nbut it is turned on only if memcache_use_advanced_pool\u003dTrue.\n\nThis patch is switching to more elaborated oslo.cache CachePool\nand adding deprecation warning about eventlet-unsafe variant\nof keystonemiddleware\u0027s memcache pool.\n\nHow to reproduce ?\n\nwith memcache_use_advanced_pool\u003dFalse\n\n1. Build clean ENV of openstack\n2. Deploy core projects (keystone,glance,nova,placement...)\n3. Run while true; do COMMAND FOR SERVICE; done\n   - several bashes, in parallel (5-7)\n\nCOMMAND FOR SERVICE:\n- openstack network list\n- openstack volume list\n- openstack server list\n- openstack image list\n\n4. Check memcached connections (which will grow up):\n    - ss | grep 11211 | wc -l   every second\n\nHow to fix and test it ?\n\nRepeat above, to fix:\n - with memcache_use_advanced_pool\u003dTrue\n   OR\n - apply this patch\n\nCompare measurements in graph.\n\n[1] https://bugs.launchpad.net/keystonemiddleware/+bug/1892852\n[2] https://bugs.launchpad.net/oslo.cache/+bug/1888394\n[3] https://bugs.launchpad.net/keystonemiddleware/+bug/1883659\n\n[4] https://review.opendev.org/c/openstack/oslo.cache/+/742193\n\n[5] https://review.opendev.org/c/openstack/keystonemiddleware/+/268664\n\nCloses-Bug: #1883659\nCloses-Bug: #1892852\nCloses-Bug: #1888394\n\nChange-Id: I0e96334b65a0bf369ebf1d88651d13feb8d2ecac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/6b64b0841b385e18f8f889fa9f23a3e7a2edb0c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/6b64b0841b385e18f8f889fa9f23a3e7a2edb0c1"}]},"branch":"refs/heads/stable/ussuri","description":"Rebase"},"3ae39554c32c195a1549f4079e2ea1224356003c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-03-24 19:50:38.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/54/777554/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/54/777554/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/54/777554/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/54/777554/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/54/777554/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/54/777554/3"}}},"commit":{"parents":[{"commit":"94c24a427a6b680c00e7bcb1940a6e9d7c9452a7","subject":"Updating lower-constraints job as non voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/94c24a427a6b680c00e7bcb1940a6e9d7c9452a7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-02-10 13:47:17.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-03-24 19:49:58.000000000","tz":60},"subject":"Switch to eventlet-safe oslo.cache\u0027s MemcacheClientPool","message":"Switch to eventlet-safe oslo.cache\u0027s MemcacheClientPool\n\nIn past days there were discussions about various issues\nwith memcached connections [1][2][3].\n\nAfter investigation it looks like common root cause for above\nproblems is keystonemiddleware. More precisely said the way\nhow keystonemiddleware is caching tokens.\n\nCurrently it\u0027s using some home-made CachePool with direct\nusage of memcached library, moreover it looks like its\napproach is not eventlet-safe.\nDiscussion can be mainly found in [4].\n\nFortunately keystonemiddleware can use \"advanced cache pool\",\nwhich is oslo.cache\u0027s implementation and was added long time ago [5],\nbut it is turned on only if memcache_use_advanced_pool\u003dTrue.\n\nThis patch is switching to more elaborated oslo.cache CachePool\nand adding deprecation warning about eventlet-unsafe variant\nof keystonemiddleware\u0027s memcache pool.\n\nHow to reproduce ?\n\nwith memcache_use_advanced_pool\u003dFalse\n\n1. Build clean ENV of openstack\n2. Deploy core projects (keystone,glance,nova,placement...)\n3. Run while true; do COMMAND FOR SERVICE; done\n   - several bashes, in parallel (5-7)\n\nCOMMAND FOR SERVICE:\n- openstack network list\n- openstack volume list\n- openstack server list\n- openstack image list\n\n4. Check memcached connections (which will grow up):\n    - ss | grep 11211 | wc -l   every second\n\nHow to fix and test it ?\n\nRepeat above, to fix:\n - with memcache_use_advanced_pool\u003dTrue\n   OR\n - apply this patch\n\nCompare measurements in graph.\n\n[1] https://bugs.launchpad.net/keystonemiddleware/+bug/1892852\n[2] https://bugs.launchpad.net/oslo.cache/+bug/1888394\n[3] https://bugs.launchpad.net/keystonemiddleware/+bug/1883659\n\n[4] https://review.opendev.org/c/openstack/oslo.cache/+/742193\n\n[5] https://review.opendev.org/c/openstack/keystonemiddleware/+/268664\n\nCloses-Bug: #1883659\nCloses-Bug: #1892852\nCloses-Bug: #1888394\n\nChange-Id: I0e96334b65a0bf369ebf1d88651d13feb8d2ecac\n(cherry picked from commit 788d3c4969e3446778496f3a9055f654602ae2c1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/3ae39554c32c195a1549f4079e2ea1224356003c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/3ae39554c32c195a1549f4079e2ea1224356003c"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
