)]}'
{"id":"openstack%2Fkeystonemiddleware~850915","triplet_id":"openstack%2Fkeystonemiddleware~stable%2Fyoga~Ifb9dc54424256d5cffe557894b2d26b2401ebf92","project":"openstack/keystonemiddleware","branch":"stable/yoga","topic":"drop-l-c-stable/yoga","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-09-11 04:25:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifb9dc54424256d5cffe557894b2d26b2401ebf92","subject":"Drop lower-constraints.txt and its testing","status":"MERGED","created":"2022-07-25 19:20:14.000000000","updated":"2022-09-11 04:27:50.000000000","submitted":"2022-09-11 04:25:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850915-drop-l-c-stable/yoga","cherry_pick_of_change":840065,"cherry_pick_of_patch_set":1,"meta_rev_id":"5caade4cadf1984197c7f8da70db33ba0002922c","_number":850915,"virtual_id_number":850915,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-11 04:25:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2022-09-09 18:01:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-11 01:14:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2022-09-09 18:01:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-25 19:20:14.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2022-07-25 20:32:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-11 01:14:43.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"6a82d4bacd1b8a996ba30561ff720cc119dd47b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-07-25 19:20:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f9cc3991b5398adfbe67d815554f288fcb1dba8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 20:32:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c351c1a84fc84a84920c887ef5a4cf72\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03a086aa213a41588497a8430845a536 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1526e4c5652c494c944deba636d40baa : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/241f8b250f6c4c98a70b4c542216465e : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85650eb88d414aaabbe424d6f30b56a1 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a76e49060794b01aa66a9985dc3c5b0 : SUCCESS in 5m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d8aebb3ea33845fd96a086bfeec90a2a : SUCCESS in 2m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0820ed8511df4b2ba2213c8389b00a96 : SUCCESS in 1h 10m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4673d239ba442caa45001ae17118e42 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f0d0c7c0c1874680ea47aae4f2fa0cc551e2c016","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-09-09 18:01:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"50b9f76a7074720951a1e14cdff07293e9716237","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 18:01:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"305aa5afce731e54b7124f9259c247a1adf89fa1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 18:25:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/81bfc9d5e0e344ff89d2b9f831c17991\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b8b090c61e74d56993116ad0a2e97ea : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c44e3a0b05fb4a4bbbaa39b4b33db1fc : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c98f8897d4c49e4a720274850a1731b : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34a24fcafb1444eea17acc0df79834d0 : SUCCESS in 5m 26s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3cb455fa875148f2bacfee3b58a40e18 : SUCCESS in 2m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/16f7191dd16548479d40d3090691b4c5 : FAILURE in 21m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cc5621730f2464299c7f8ab87fbd18c : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":1},{"id":"24bec50cb6c463ce9652830b95e4e3d6168e6f1a","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-09-11 01:14:43.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3159fc629b24c2cfc9d7d7e16bedb0deb7b0325f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-11 02:52:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d39864a32ea4ef68b7d2dbf9e6c9625\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84198d4a0df340eaac59961620689e27 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/051b8bdf9b6e4e3ba9ca19f93b8207a6 : SUCCESS in 2m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d426411220814a52b7b09ddc13e59e01 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2385bea53c84276854ca582261bdc1e : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2589536ec03c42aca4d6d7084265a1ff : SUCCESS in 6m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c10beb90b409439ab1ae0001f7337b27 : SUCCESS in 2m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4adb802b4b64ffdafb112ee4cf87409 : SUCCESS in 1h 36m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1062350b1c6449e890d3785de95c9def : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"06db35e4ac97af732096e2d798ef2d80e43c0055","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-11 02:53:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f813be8272464321020b308292720da33141e3dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-11 04:25:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e41887a18b264f88b9bd2cc2cae3a105\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45b0a586fd8241179828cca3f3dc28be : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60352f9ff6a4477b8090e9bef412b547 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06ca9ce61d664ec8897d3e2520f9e528 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9345e9142ae46b5b5da8cc9aaede295 : SUCCESS in 5m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e6fb66f328ad4b3cae38f7e0e0ef303b : SUCCESS in 3m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa1969221d2942d992ea31a015247be2 : SUCCESS in 1h 30m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/790469ebf6474a719eedacb3a1f25ec9 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":1},{"id":"aa169e539f68b5a96b30b176757b2c0e0c9160f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-11 04:25:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5caade4cadf1984197c7f8da70db33ba0002922c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-11 04:27:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e09589bbe9714f13a91e384e583f23e8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/458e97fc8e4147da91bf4afe316c00be : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87621acc6a62420486a1ec736e4acc4b : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"67756adbb9cb7210fda08a2a40172a78699d0d09","revisions":{"67756adbb9cb7210fda08a2a40172a78699d0d09":{"kind":"REWORK","_number":1,"created":"2022-07-25 19:20:14.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/15/850915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/15/850915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/15/850915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/15/850915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/15/850915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/15/850915/1"}}},"commit":{"parents":[{"commit":"4f9ed4f7dd1b8d26df83830215b2065df9a27e7b","subject":"Update TOX_CONSTRAINTS_FILE for stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/4f9ed4f7dd1b8d26df83830215b2065df9a27e7b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-04-30 22:15:11.000000000","tz":-300},"committer":{"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","date":"2022-07-25 19:20:14.000000000","tz":0},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: Ifb9dc54424256d5cffe557894b2d26b2401ebf92\n(cherry picked from commit d60ccd46b9d11a3127f20dd6da69d2faedb50269)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/67756adbb9cb7210fda08a2a40172a78699d0d09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/67756adbb9cb7210fda08a2a40172a78699d0d09"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[]}
