)]}'
{"id":"openstack%2Fkeystonemiddleware~850973","triplet_id":"openstack%2Fkeystonemiddleware~master~Ifbafd2d9ab5d7075276685a70e8394db0d75050c","project":"openstack/keystonemiddleware","branch":"master","attention_set":{},"removed_from_attention_set":{"35058":{"account":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"last_update":"2024-09-06 16:11:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifbafd2d9ab5d7075276685a70e8394db0d75050c","subject":"Use TOX_CONSTRAINTS_FILE","status":"MERGED","created":"2022-07-26 06:12:09.000000000","updated":"2024-09-06 16:12:33.000000000","submitted":"2024-09-06 16:11:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850973","meta_rev_id":"9af2b54bdfc1f938b31086f6bb081bbedb964060","_number":850973,"virtual_id_number":850973,"owner":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-06 16:11:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"recommended":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"all":[{"value":1,"date":"2024-08-28 12:16:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2024-09-06 14:22:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2024-08-30 14:42:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":1,"date":"2024-09-06 14:22:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-26 07:47:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-29 08:05:47.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-28 12:16:06.000000000","updated_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"reviewer":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"state":"REVIEWER"},{"updated":"2024-08-30 14:42:35.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-09-06 14:22:00.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"484c8f0e6d3c7ac060ce47bfbac346cdb87f77ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-07-26 06:12:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"624a76b140c476b749d43a1041c8f2f26907edeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 07:47:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/466e015f24bc4453af7b857f1952d8f1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74cbc424071c4a388f48a00264f93db8 : FAILURE in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db6cd38c572043c3a0a853d969f8e8b8 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/108b4db77f7e4a7e96dc9a1fdfe523eb : FAILURE in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26fc5858ca8444d19d24738f419f28fd : FAILURE in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3577f4daa8143ba8d2ee3f07209a703 : FAILURE in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9e2f582588946f8b2af3b0620eaeaf5 : SUCCESS in 7m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ed0e2c6b3f44c9e87535696890c3e61 : FAILURE in 1h 26m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/077ad28056514315ab0170298fbaf14f : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9621eb02c5ecf347c343101eae306103df619c9b","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-07-26 09:11:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b1e156c0e1820e9583ef37e1d573c5c54935c9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 10:52:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/156e07c70d49422cb99d418b1e37c38d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e961d4301de49b88a6be8cf11ef14c2 : FAILURE in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c248d7aa1dd45039706e39caa81838b : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c5c31a6cc1d4119b191fb95fa9727e7 : FAILURE in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0a29b107cbd4708a6ab9e141d233bb7 : FAILURE in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd850bfcd31d4fdf992a52f240e74379 : FAILURE in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3a7c740c87b44399ef3f81eabd294b6 : SUCCESS in 6m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a118cb4005ea487daed3ee27177e3ee3 : SUCCESS in 1h 39m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f614ea74d8364f90aafd8c1c40989d04 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"362155bbe8e509e2eb84f8ff29da881a616f7fec","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-07-27 08:27:27.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"ee712fe73a0eb81a02137e501c605b171fe81442","tag":"autogenerated:gerrit:restore","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2023-08-29 08:05:24.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"caad4df0e984fceaef68b4cd316b076a80093781","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 08:05:40.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/keystonemiddleware for 850973,1","accounts_in_message":[],"_revision_number":1},{"id":"f9b3db85b7e12777d66ea48633aba97a598487d5","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-29 08:05:47.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/keystonemiddleware for 850973,1","accounts_in_message":[],"_revision_number":1},{"id":"569c8e157f05f0521d9129ba8b5cad006b6125e6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2023-08-29 08:06:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"8f54cdc0e4a20843f6c7189b27df546ef049ccf7","tag":"autogenerated:gerrit:restore","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2023-08-29 08:06:50.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"136f931931f7a9171e156faaa33b93e9b8fb5f98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 08:07:06.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/keystonemiddleware for 850973,1","accounts_in_message":[],"_revision_number":1},{"id":"feee20a26cf2ce2c1eb2393c868a7787700c5bd0","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-08-29 08:07:08.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/keystonemiddleware for 850973,1","accounts_in_message":[],"_revision_number":1},{"id":"2ff7e4421d63b5cc639b76ee3b802f5e6b3db258","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2023-08-29 08:15:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nThe following files contain Git conflicts:\n* tox.ini\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f820832bb84b32a7c4234d1aa178575895e05e9a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2023-08-29 08:32:12.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"3abf505b36b9b64bd5ca4a272e385da7237c514f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2023-08-29 08:37:22.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"541427277b7469627aafd23f2425d3dcd4c85cb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 10:27:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bd30d0cebff47b089429d255683f598\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41a35e2910ad44cb9c035616e347a211 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/861b1b567d8344539065553c85eada96 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75a33b0f1c16422ab8075441362e5953 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09977382d109464d89db0535fd957e31 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93dcdd74377448ec9af43e2f24a98041 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d174dbadab94df0aaaed62147794cad : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca410991be8248208cd0fe7e86fb4b51 : SUCCESS in 6m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/264e8196b5ae46d0a2c7bab27df98db0 : SUCCESS in 1h 42m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf9187e1b2bf4892b6a07209ed61ea38 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":3},{"id":"ff3c8730661c2c42ac24775ea454fc0922153a59","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2024-08-28 12:16:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ede9032780dbcd95feb7560ffb89b39ce557f0be","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-08-30 14:42:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3e5288871d026942f9880260968e9e123294b8fd","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-09-06 14:22:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"561afa118d241b155a4440ac6af7bbc1d652ba40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 14:22:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f3d7c87dfab9f958f0d563d2f8791bdec570fc30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 16:11:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e49fdeab8eeb43b5acb23392e7b2c2a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc49c0e58f654357b916a32ee547dfe4 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db0c3a7fc2bb4e4c9d955d804b3bfb15 : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c3513ab8670640a787197ffe023a75f2 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37156fd715034870b6570a2bf6e80a72 : SUCCESS in 5m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/193955492a724a3a9285c92490966658 : SUCCESS in 1h 39m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6776dffd47c14e7b81450ba670f4e95c : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":3},{"id":"67925be2838bed6d8566a142b2a76ad4f9005627","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 16:11:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9af2b54bdfc1f938b31086f6bb081bbedb964060","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 16:12:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8fb18bf71044997a47d9eba1f7ad853\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8642b687a10e449885c4f561095534c8 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9079e9e5e294f88a0b767f549a2479f : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"10134908b44e21888f293e636c29b8f83aad61f8","revisions":{"18157d842bfcba81d3524c18550a9aa4baa3c712":{"kind":"REWORK","_number":1,"created":"2022-07-26 06:12:09.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/73/850973/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/73/850973/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/73/850973/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/73/850973/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/73/850973/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/73/850973/1"}}},"commit":{"parents":[{"commit":"2bda844bb219df355d74b5c5b21f86244921a1c2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2bda844bb219df355d74b5c5b21f86244921a1c2"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-26 06:10:42.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-26 06:10:55.000000000","tz":480},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\nThis allows to use upper-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\nChange-Id: Ifbafd2d9ab5d7075276685a70e8394db0d75050c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/18157d842bfcba81d3524c18550a9aa4baa3c712"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/18157d842bfcba81d3524c18550a9aa4baa3c712"}]},"branch":"refs/heads/master"},"5249c12c4597510099a501f14e57bd955f8dc32d":{"kind":"REWORK","_number":2,"created":"2023-08-29 08:15:06.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/73/850973/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/73/850973/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/73/850973/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/73/850973/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/73/850973/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/73/850973/2"}}},"commit":{"parents":[{"commit":"d36c86ce4c8822de63a436cf109baddb313af40f","subject":"Merge \"Switch to 2023.1 Python3 unit tests and generic template name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d36c86ce4c8822de63a436cf109baddb313af40f"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-26 06:10:42.000000000","tz":480},"committer":{"name":"Ke Niu","email":"niuke@inspur.com","date":"2023-08-29 08:15:06.000000000","tz":0},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\nThis allows to use upper-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\nChange-Id: Ifbafd2d9ab5d7075276685a70e8394db0d75050c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5249c12c4597510099a501f14e57bd955f8dc32d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/5249c12c4597510099a501f14e57bd955f8dc32d"}]},"branch":"refs/heads/master","description":"Rebase"},"10134908b44e21888f293e636c29b8f83aad61f8":{"kind":"REWORK","_number":3,"created":"2023-08-29 08:37:22.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/73/850973/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/73/850973/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/73/850973/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/73/850973/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/73/850973/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/73/850973/3"}}},"commit":{"parents":[{"commit":"d36c86ce4c8822de63a436cf109baddb313af40f","subject":"Merge \"Switch to 2023.1 Python3 unit tests and generic template name\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/d36c86ce4c8822de63a436cf109baddb313af40f"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-26 06:10:42.000000000","tz":480},"committer":{"name":"Ke Niu","email":"niuke@inspur.com","date":"2023-08-29 08:37:21.000000000","tz":0},"subject":"Use TOX_CONSTRAINTS_FILE","message":"Use TOX_CONSTRAINTS_FILE\n\nUPPER_CONSTRAINTS_FILE is old name and deprecated\nThis allows to use upper-constraints file as more\nreadable way instead of UPPER_CONSTRAINTS_FILE\u003d\u003clower-constraints file\u003e.\n\nChange-Id: Ifbafd2d9ab5d7075276685a70e8394db0d75050c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/10134908b44e21888f293e636c29b8f83aad61f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/10134908b44e21888f293e636c29b8f83aad61f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
