)]}'
{"id":"openstack%2Fkeystonemiddleware~851319","triplet_id":"openstack%2Fkeystonemiddleware~master~Ia8b0d130ba376f396cc6501c26c5d7f3088e5661","project":"openstack/keystonemiddleware","branch":"master","topic":"bug/1982988","attention_set":{"30736":{"account":{"_account_id":30736,"name":"HYSong","email":"g.terry.song@gmail.com","username":"Shy"},"last_update":"2022-07-28 08:44:57.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ia8b0d130ba376f396cc6501c26c5d7f3088e5661","subject":"fix unkonown uuid","status":"NEW","created":"2022-07-28 07:05:57.000000000","updated":"2022-07-28 08:44:57.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a8827127528ff9bf73491b12249b3cc7860de04e","_number":851319,"virtual_id_number":851319,"owner":{"_account_id":30736,"name":"HYSong","email":"g.terry.song@gmail.com","username":"Shy"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-28 08:44:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-28 08:44:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"491a3bcfa8d6605dc73fbc3387c5b8f3ddb8cf9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30736,"name":"HYSong","email":"g.terry.song@gmail.com","username":"Shy"},"date":"2022-07-28 07:05:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8827127528ff9bf73491b12249b3cc7860de04e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 08:44:57.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5673b3d9ed3c4d2f849145edf26c6013\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b09dcbf8d37d4611bdb402e4ee7200a1 : FAILURE in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67b61890cade42dbbb1cefe823ae9091 : FAILURE in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6e0d5e8104b4855b0291e2388c74a8c : FAILURE in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8bead5dfadb6425caacdf5ba25ca1b5e : FAILURE in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5f229853e2a4a6d9e7f1b62df98e58b : FAILURE in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/747cfd8df0764d8c8a8752861144a28c : SUCCESS in 8m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb4d7c00a27844439e85a8ad26d5dc07 : SUCCESS in 1h 36m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c5109e2732535ec54e3de43fc4de808e1cbd389d","revisions":{"c5109e2732535ec54e3de43fc4de808e1cbd389d":{"kind":"REWORK","_number":1,"created":"2022-07-28 07:05:57.000000000","uploader":{"_account_id":30736,"name":"HYSong","email":"g.terry.song@gmail.com","username":"Shy"},"ref":"refs/changes/19/851319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/19/851319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/19/851319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/19/851319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/19/851319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/19/851319/1"}}},"commit":{"parents":[{"commit":"2bda844bb219df355d74b5c5b21f86244921a1c2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2bda844bb219df355d74b5c5b21f86244921a1c2"}]}],"author":{"name":"Shy","email":"g.terry.song@gmail.com","date":"2022-07-28 06:33:53.000000000","tz":480},"committer":{"name":"Shy","email":"g.terry.song@gmail.com","date":"2022-07-28 07:03:15.000000000","tz":480},"subject":"fix unkonown uuid","message":"fix unkonown uuid\n\nThis change fix bug:\nhttps://bugs.launchpad.net/keystone/+bug/1982988\n\nThe param of uuid is initailize with zero instead of unknown.\n\nClose-Bug: #1982988\nChange-Id: Ia8b0d130ba376f396cc6501c26c5d7f3088e5661\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c5109e2732535ec54e3de43fc4de808e1cbd389d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/c5109e2732535ec54e3de43fc4de808e1cbd389d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
