)]}'
{"id":"openstack%2Fkeystonemiddleware~852590","triplet_id":"openstack%2Fkeystonemiddleware~master~I32f9dc596525e912e37984764f68564e26ecfd3b","project":"openstack/keystonemiddleware","branch":"master","topic":"fix-ut","attention_set":{},"removed_from_attention_set":{"33920":{"account":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"last_update":"2022-08-16 17:00:39.000000000","reason":"Change was submitted"},"31857":{"account":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"last_update":"2022-08-16 17:00:39.000000000","reason":"Change was submitted"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2022-08-16 15:12:42.000000000","reason":"\u003cGERRIT_ACCOUNT_7973\u003e replied on the change","reason_account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},"33455":{"account":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"last_update":"2022-08-10 05:34:01.000000000","reason":"\u003cGERRIT_ACCOUNT_33455\u003e replied on the change","reason_account":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}}},"hashtags":[],"change_id":"I32f9dc596525e912e37984764f68564e26ecfd3b","subject":"Fix logging notifier unit test","status":"MERGED","created":"2022-08-09 14:38:23.000000000","updated":"2022-08-16 17:01:56.000000000","submitted":"2022-08-16 17:00:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852590-fix-ut","meta_rev_id":"696ead84f1f076d45933d502dde2aab6d25d3aa0","_number":852590,"virtual_id_number":852590,"owner":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-16 17:00:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"all":[{"value":2,"date":"2022-08-09 16:20:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2022-08-16 15:12:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-10 05:34:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":1,"date":"2022-08-16 15:12:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"}],"CC":[{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-09 16:11:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-09 16:20:33.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2022-08-09 16:47:37.000000000","updated_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"reviewer":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"state":"REVIEWER"},{"updated":"2022-08-09 16:47:37.000000000","updated_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"reviewer":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"state":"REVIEWER"},{"updated":"2022-08-09 16:47:37.000000000","updated_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2022-08-09 16:48:34.000000000","updated_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"reviewer":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"state":"CC"},{"updated":"2022-08-09 16:48:34.000000000","updated_by":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"reviewer":{"_account_id":33920,"name":"Yusuke Niimi","email":"niimi.yusuke@fujitsu.com","username":"yniimi"},"state":"CC"},{"updated":"2022-08-10 05:34:01.000000000","updated_by":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"reviewer":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"state":"REVIEWER"}],"messages":[{"id":"e3e09d3a35b86f51438b4fa92d563f6dc4c799be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"date":"2022-08-09 14:38:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"94253a87fc013fa186fa4ebc8c9a133b69537670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 16:11:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46f7846ad86b444b9ec4dbae972f4378\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5be1e1ecaf3b4d558aa3c6ffca805ebe : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb62822ac332466f93b617a010eb5fca : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd2eba46e3f944a2b83b91bfc35acedb : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4e42620d0ee4e019605696ccd611d83 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cdca999fe0d94485af1dcd4d46ce156f : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a9aadca58c54dc9a319c0435a0c9269 : SUCCESS in 4m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02155f3a6fb8487684f1dcd202bb654a : SUCCESS in 1h 28m 33s","accounts_in_message":[],"_revision_number":1},{"id":"87eb2f222b802ba7ce8ed6fa98eb030e7146e1f4","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2022-08-09 16:20:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2d2fc07943df032c2d056c54d77600dfe1c3d429","author":{"_account_id":33455,"name":"Hiromu Asahina","email":"hiromu.a5a@gmail.com","username":"h_asahina"},"date":"2022-08-10 05:34:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ef159f8d1c56aef3d2f98178bb1dcfe5f3cc748f","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2022-08-16 15:12:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c3acbb3072d3da744a0640e6d97d8bd1f09a44e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 15:13:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca1b6ac90ac515de305e7ea8c7fd7acb4eb4426d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 17:00:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/896d9d10840a4066a62daca0d28562f0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83611be2ecd04198b64975ce39c27e09 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc0ace314da744b8b0a8c241b3a233a8 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f9d6719095c4334b3e237143353a5c9 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf33ed7724f84ee18ee3073de19a0173 : SUCCESS in 5m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6f73acbc32d4ff9aba356305d5b63a6 : SUCCESS in 1h 41m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e9379f22cb8ee34fe107c063ccb42f4fed6121d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 17:00:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"696ead84f1f076d45933d502dde2aab6d25d3aa0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 17:01:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef5465d6a4dc46bba431abeacbbee6a9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f81ee13fbf3b49da81b5fc09d905cd34 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c49da6887eaff67d19fe0d12fb1046dde3bdaec","revisions":{"9c49da6887eaff67d19fe0d12fb1046dde3bdaec":{"kind":"REWORK","_number":1,"created":"2022-08-09 14:38:23.000000000","uploader":{"_account_id":31857,"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","username":"ueha"},"ref":"refs/changes/90/852590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystonemiddleware","ref":"refs/changes/90/852590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/90/852590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/90/852590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystonemiddleware refs/changes/90/852590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystonemiddleware refs/changes/90/852590/1"}}},"commit":{"parents":[{"commit":"2bda844bb219df355d74b5c5b21f86244921a1c2","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/2bda844bb219df355d74b5c5b21f86244921a1c2"}]}],"author":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2022-08-09 14:07:21.000000000","tz":0},"committer":{"name":"Ayumu Ueha","email":"ueha.ayumu@fujitsu.com","date":"2022-08-09 14:33:32.000000000","tz":0},"subject":"Fix logging notifier unit test","message":"Fix logging notifier unit test\n\nFor unknown reasons, the `create_notifier()` in\n`test_api_request_no_messaging()` used `oslo_messaging.Notifier`\ninstead of the `_LogNotifier` that should be originally used,\ncausing unit test to fail.\n\nThis patch fixes this issue by changing `use_oslo_messaging` config to\nFalse for this test.\n\nChange-Id: I32f9dc596525e912e37984764f68564e26ecfd3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/9c49da6887eaff67d19fe0d12fb1046dde3bdaec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystonemiddleware/commit/9c49da6887eaff67d19fe0d12fb1046dde3bdaec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[]}
